Skip to content

Makefile: fix api generation target - #67

Merged
openshift-merge-robot merged 1 commit into
openshift:mainfrom
ironcladlou:fix-api-target
Feb 24, 2021
Merged

Makefile: fix api generation target#67
openshift-merge-robot merged 1 commit into
openshift:mainfrom
ironcladlou:fix-api-target

Conversation

@ironcladlou

Copy link
Copy Markdown
Contributor

This patch fixes an api generation regression introduced by #60
which causes the new thirdparty package contents to be processed by the
code generation tooling. The codegen tools should only be applied to the
hypershift api package itself, not things in thirdparty.

This patch fixes an api generation regression introduced by openshift#60
which causes the new thirdparty package contents to be processed by the
code generation tooling. The codegen tools should only be applied to the
hypershift api package itself, not things in thirdparty.
@enxebre

enxebre commented Feb 24, 2021

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ironcladlou

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 24, 2021
@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Feb 24, 2021
@openshift-merge-robot
openshift-merge-robot merged commit 5442e03 into openshift:main Feb 24, 2021
@ironcladlou ironcladlou mentioned this pull request Feb 25, 2021
4 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants