Skip to content

[release-4.14] OCPBUGS-29000,OCPBUGS-28984,OCPBUGS-28985,OCPBUGS-28986,OCPBUGS-19956: Support Disconnected HCP - #3520

Merged
openshift-merge-bot[bot] merged 4 commits into
openshift:release-4.14from
bryan-cox:support-disconnected-hcp-4.14
Feb 7, 2024
Merged

[release-4.14] OCPBUGS-29000,OCPBUGS-28984,OCPBUGS-28985,OCPBUGS-28986,OCPBUGS-19956: Support Disconnected HCP#3520
openshift-merge-bot[bot] merged 4 commits into
openshift:release-4.14from
bryan-cox:support-disconnected-hcp-4.14

Conversation

@bryan-cox

@bryan-cox bryan-cox commented Feb 5, 2024

Copy link
Copy Markdown
Member

What this PR does / why we need it:

Which issue(s) this PR fixes
Manual cherry pick

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

OCPBUGS-11939: Use the image metadata from any ICSP/IDMS in a management
 cluster

Signed-off-by: Bryan Cox <brcox@redhat.com>
Adds functionality to support extracting image files into
release-manifests/image-reference config directory when working in a
disconnected environment in the ignition server. Also, now catches the
error when extracting image references to the config directory.

Signed-off-by: Bryan Cox <brcox@redhat.com>
Include the OpenShiftTrustedCABundle ConfigMap in the ignition server
deployment.

Signed-off-by: Bryan Cox <brcox@redhat.com>
Adds functionality to support extracting image files into
release-manifests/image-reference config directory when working in a
disconnected environment in the ignition server. Also, now catches the
error when extracting image references to the config directory.

Signed-off-by: Bryan Cox <brcox@redhat.com>
@openshift-ci-robot openshift-ci-robot added jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. labels Feb 5, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@bryan-cox: This pull request references Jira Issue OCPBUGS-29000, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-11939 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

The bug has been updated to refer to the pull request using the external bug tracker.

This pull request references Jira Issue OCPBUGS-28984, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18128 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

The bug has been updated to refer to the pull request using the external bug tracker.

This pull request references Jira Issue OCPBUGS-28985, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18460 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

The bug has been updated to refer to the pull request using the external bug tracker.

This pull request references Jira Issue OCPBUGS-28986, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18602 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

The bug has been updated to refer to the pull request using the external bug tracker.

This pull request references Jira Issue OCPBUGS-19956, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18879 is in the state Closed (Done), which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @elsapassaro

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

What this PR does / why we need it:

Which issue(s) this PR fixes
Manual cherry pick

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added do-not-merge/needs-area do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. labels Feb 5, 2024
@openshift-ci

openshift-ci Bot commented Feb 5, 2024

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@bryan-cox

Copy link
Copy Markdown
Member Author

This is a manual pick of #3064 which needed to be rebased.

@bryan-cox

Copy link
Copy Markdown
Member Author

/test images
/test verify
/test unit
/test e2e-aws

@openshift-ci-robot

Copy link
Copy Markdown

@bryan-cox: This pull request references Jira Issue OCPBUGS-29000, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-11939 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-28984, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18128 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-28985, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18460 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-28986, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18602 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-19956, which is valid.

4 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18879 is in the state Closed (Done), which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • bug has dependents

Requesting review from QA contact:
/cc @elsapassaro

Details

In response to this:

What this PR does / why we need it:

Which issue(s) this PR fixes
Manual cherry pick

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Feb 5, 2024

Copy link
Copy Markdown
Contributor

@bryan-cox: No presubmit jobs available for openshift/hypershift@rhtap/references/release-4.14

Details

In response to this:

/test images
/test verify
/test unit
/test e2e-aws

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@bryan-cox
bryan-cox marked this pull request as ready for review February 5, 2024 14:32
@bryan-cox
bryan-cox changed the base branch from rhtap/references/release-4.14 to release-4.14 February 5, 2024 14:33
@openshift-ci-robot

Copy link
Copy Markdown

@bryan-cox: This pull request references Jira Issue OCPBUGS-29000, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-11939 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-11939 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-28984, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18128 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18128 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-28985, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18460 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18460 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-28986, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18602 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18602 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @LiangquanLi930

This pull request references Jira Issue OCPBUGS-19956, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18879 is in the state Closed (Done), which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18879 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @elsapassaro

Details

In response to this:

What this PR does / why we need it:

Which issue(s) this PR fixes
Manual cherry pick

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added area/control-plane-operator Indicates the PR includes changes for the control plane operator - in an OCP release area/hypershift-operator Indicates the PR includes changes for the hypershift operator and API - outside an OCP release and removed do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. do-not-merge/needs-area labels Feb 5, 2024
@openshift-ci
openshift-ci Bot requested review from enxebre and sjenning February 5, 2024 14:41
@jparrill

jparrill commented Feb 5, 2024

Copy link
Copy Markdown
Contributor

/approve
/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 5, 2024
@bryan-cox

Copy link
Copy Markdown
Member Author

/test e2e-aws

@LiangquanLi930

Copy link
Copy Markdown
Contributor

/lgtm

@sjenning

sjenning commented Feb 6, 2024

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci

openshift-ci Bot commented Feb 6, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bryan-cox, jparrill, sjenning

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 6, 2024
@LiangquanLi930

Copy link
Copy Markdown
Contributor

/test e2e-kubevirt-aws-ovn

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 6d15150 and 2 for PR HEAD aabad1b in total

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD 21824b2 and 1 for PR HEAD aabad1b in total

@openshift-ci

openshift-ci Bot commented Feb 7, 2024

Copy link
Copy Markdown
Contributor

@bryan-cox: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 15533d4 into openshift:release-4.14 Feb 7, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@bryan-cox: Jira Issue OCPBUGS-29000: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-29000 has been moved to the MODIFIED state.

Jira Issue OCPBUGS-28984: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-28984 has been moved to the MODIFIED state.

Jira Issue OCPBUGS-28985: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-28985 has been moved to the MODIFIED state.

Jira Issue OCPBUGS-28986: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-28986 has been moved to the MODIFIED state.

Jira Issue OCPBUGS-19956: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-19956 has been moved to the MODIFIED state.

Details

In response to this:

What this PR does / why we need it:

Which issue(s) this PR fixes
Manual cherry pick

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@bryan-cox
bryan-cox deleted the support-disconnected-hcp-4.14 branch February 7, 2024 13:31
@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.14.0-0.nightly-2024-02-07-120144

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/control-plane-operator Indicates the PR includes changes for the control plane operator - in an OCP release area/hypershift-operator Indicates the PR includes changes for the hypershift operator and API - outside an OCP release jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants