Skip to content

OCPBUGS-7840: Untangle kas port - #2964

Merged
openshift-merge-robot merged 2 commits into
openshift:mainfrom
enxebre:untangle-kas-port
Sep 6, 2023
Merged

OCPBUGS-7840: Untangle kas port#2964
openshift-merge-robot merged 2 commits into
openshift:mainfrom
enxebre:untangle-kas-port

Conversation

@enxebre

@enxebre enxebre commented Aug 30, 2023

Copy link
Copy Markdown
Member

What this PR does / why we need it:
With this change, existing HCs defaulting Networking.APIServer.Port to 443 will preserve current behaviour of overriding the kas endpoint in the guest cluster.
This is to avoid data plane rollouts that would result from changing the haproxy config.

New HCs wont default to 443 and they won't override the kas endpoint. Setting Networking.APIServer.Port to 443 by a consumer is still techincally possible but unrecommended.

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

@openshift-ci openshift-ci Bot added area/control-plane-operator Indicates the PR includes changes for the control plane operator - in an OCP release area/hypershift-operator Indicates the PR includes changes for the hypershift operator and API - outside an OCP release labels Aug 30, 2023
@openshift-ci

openshift-ci Bot commented Aug 30, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: enxebre

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. and removed do-not-merge/needs-area labels Aug 30, 2023
@enxebre
enxebre force-pushed the untangle-kas-port branch from ec097d1 to 9e511ce Compare August 30, 2023 16:50
@enxebre

enxebre commented Aug 30, 2023

Copy link
Copy Markdown
Member Author

/test e2e-aws

@enxebre
enxebre force-pushed the untangle-kas-port branch from 9e511ce to 24d3a58 Compare August 31, 2023 10:12
@enxebre

enxebre commented Sep 1, 2023

Copy link
Copy Markdown
Member Author

/test e2e-aws

@enxebre

enxebre commented Sep 4, 2023

Copy link
Copy Markdown
Member Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 4, 2023
@netlify

netlify Bot commented Sep 4, 2023

Copy link
Copy Markdown

Deploy Preview for hypershift-docs ready!

Name Link
🔨 Latest commit 2f58b6b
🔍 Latest deploy log https://app.netlify.com/sites/hypershift-docs/deploys/64f5b04928b0ef00096f5edb
😎 Deploy Preview https://deploy-preview-2964--hypershift-docs.netlify.app/reference/api
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site configuration.

@openshift-ci openshift-ci Bot added area/ci-tooling Indicates the PR includes changes for CI or tooling area/cli Indicates the PR includes changes for CLI area/documentation Indicates the PR includes changes for documentation labels Sep 4, 2023
@enxebre
enxebre force-pushed the untangle-kas-port branch 2 times, most recently from f37ae81 to a5c952f Compare September 4, 2023 13:40
…ters

With this change, existing HCs defaulting Networking.APIServer.Port to 443 will preserve current behaviour of overriding the kas endpoint in the guest cluster.
This is to avoid rollouts that would result from changing the haproxy config.

New HCs wont default to 443 and they won't override the kas endpoint. Seeting Networking.APIServer.Port to 443 by a consumer is still techincally possible but unrecommended.
@enxebre

enxebre commented Sep 5, 2023

Copy link
Copy Markdown
Member Author

=== CONT TestNodePool/Main/TestNTOMachineConfigAppliedInPlace
nodepool_nto_machineconfig_test.go:158:
failed waiting for all pods in the NTO MachineConfig update verification DS to be ready: timed out waiting for the condition
Unexpected error:
<*errors.errorString | 0xc0002ffd10>:
timed out waiting for the condition
{
s: "timed out waiting for the condition",
}
occurred
=== RUN TestNodePool/Teardown

/e2e-kubevirt-aws-ovn

@enxebre enxebre changed the title Untangle kas port OCPBUGS-7840: Untangle kas port Sep 5, 2023
@openshift-ci-robot openshift-ci-robot added jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Sep 5, 2023
@openshift-ci-robot

Copy link
Copy Markdown

@enxebre: This pull request references Jira Issue OCPBUGS-7840, which is invalid:

  • expected the bug to target the "4.14.0" version, but no target version was set

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

What this PR does / why we need it:
With this change, existing HCs defaulting Networking.APIServer.Port to 443 will preserve current behaviour of overriding the kas endpoint in the guest cluster.
This is to avoid rollouts that would result from changing the haproxy config.

New HCs wont default to 443 and they won't override the kas endpoint. Setting Networking.APIServer.Port to 443 by a consumer is still techincally possible but unrecommended.

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@enxebre

enxebre commented Sep 5, 2023

Copy link
Copy Markdown
Member Author

/jira refresh

@openshift-ci-robot openshift-ci-robot added the jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. label Sep 5, 2023
@openshift-ci-robot

Copy link
Copy Markdown

@enxebre: This pull request references Jira Issue OCPBUGS-7840, which is valid. The bug has been moved to the POST state.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.0) matches configured target version for branch (4.14.0)
  • bug is in the state ASSIGNED, which is one of the valid states (NEW, ASSIGNED, POST)

No GitHub users were found matching the public email listed for the QA contact in Jira (jiezhao@redhat.com), skipping review request.

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot removed the jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. label Sep 5, 2023
@csrwng

csrwng commented Sep 5, 2023

Copy link
Copy Markdown
Contributor

/lgtm
/retest-required

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 5, 2023
@openshift-ci

openshift-ci Bot commented Sep 5, 2023

Copy link
Copy Markdown
Contributor

@enxebre: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-ibmcloud-iks 4a5fb87 link false /test e2e-ibmcloud-iks
ci/prow/e2e-ibmcloud-roks 4a5fb87 link false /test e2e-ibmcloud-roks

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-ci-robot

Copy link
Copy Markdown

@enxebre: This pull request references Jira Issue OCPBUGS-7840, which is valid.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.0) matches configured target version for branch (4.14.0)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)

No GitHub users were found matching the public email listed for the QA contact in Jira (jiezhao@redhat.com), skipping review request.

Details

In response to this:

What this PR does / why we need it:
With this change, existing HCs defaulting Networking.APIServer.Port to 443 will preserve current behaviour of overriding the kas endpoint in the guest cluster.
This is to avoid data plane rollouts that would result from changing the haproxy config.

New HCs wont default to 443 and they won't override the kas endpoint. Setting Networking.APIServer.Port to 443 by a consumer is still techincally possible but unrecommended.

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@enxebre

enxebre commented Sep 6, 2023

Copy link
Copy Markdown
Member Author

/test e2e-kubevirt-aws-ovn

@enxebre

enxebre commented Sep 6, 2023

Copy link
Copy Markdown
Member Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 6, 2023
@openshift-merge-robot
openshift-merge-robot merged commit 8c24b8c into openshift:main Sep 6, 2023
@openshift-ci-robot

Copy link
Copy Markdown

@enxebre: Jira Issue OCPBUGS-7840: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-7840 has been moved to the MODIFIED state.

Details

In response to this:

What this PR does / why we need it:
With this change, existing HCs defaulting Networking.APIServer.Port to 443 will preserve current behaviour of overriding the kas endpoint in the guest cluster.
This is to avoid data plane rollouts that would result from changing the haproxy config.

New HCs wont default to 443 and they won't override the kas endpoint. Setting Networking.APIServer.Port to 443 by a consumer is still techincally possible but unrecommended.

Which issue(s) this PR fixes (optional, use fixes #<issue_number>(, fixes #<issue_number>, ...) format, where issue_number might be a GitHub issue, or a Jira story:
Fixes #

Checklist

  • Subject and description added to both, commit and PR.
  • Relevant issues have been referenced.
  • This change includes docs.
  • This change includes unit tests.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.14.0-0.nightly-2023-09-11-201102

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.14.0-0.nightly-2023-09-12-024050

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.14.0-0.nightly-2023-09-15-101929

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/ci-tooling Indicates the PR includes changes for CI or tooling area/cli Indicates the PR includes changes for CLI area/control-plane-operator Indicates the PR includes changes for the control plane operator - in an OCP release area/documentation Indicates the PR includes changes for documentation area/hypershift-operator Indicates the PR includes changes for the hypershift operator and API - outside an OCP release jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants