OCPBUGS-11939, OCPBUGS-18128, OCPBUGS-18460 , OCPBUGS-18602, OCPBUGS-18879: Support Disconnected HCP - #2950
Conversation
|
Skipping CI for Draft Pull Request. |
7663a9d to
21bb0e1
Compare
|
Would we want to add hypershift/hypershift-operator/main.go Line 350 in 21bb0e1 |
|
@bryan-cox: This pull request references Jira Issue OCPBUGS-11939, which is valid. The bug has been moved to the POST state. 3 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (lshilin@redhat.com), skipping review request. The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@bryan-cox: This pull request references Jira Issue OCPBUGS-11939, which is valid. 3 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (lshilin@redhat.com), skipping review request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@bryan-cox: This pull request references Jira Issue OCPBUGS-11939, which is valid. 3 validation(s) were run on this bug
No GitHub users were found matching the public email listed for the QA contact in Jira (lshilin@redhat.com), skipping review request. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
db43d51 to
3488a43
Compare
|
@bryan-cox: This pull request references Jira Issue OCPBUGS-18128, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
edb01fd to
adbe5af
Compare
Include the OpenShiftTrustedCABundle ConfigMap in the ignition server deployment. Signed-off-by: Bryan Cox <brcox@redhat.com>
Include the OpenShiftTrustedCABundle ConfigMap in the HCCO deployment. Signed-off-by: Bryan Cox <brcox@redhat.com>
Update the HCCO release provider to include registry overrides and OpenShift image registry overrides. Signed-off-by: Bryan Cox <brcox@redhat.com>
Adds functionality to support extracting image files into release-manifests/image-reference config directory when working in a disconnected environment in the ignition server. Also, now catches the error when extracting image references to the config directory. Signed-off-by: Bryan Cox <brcox@redhat.com>
1d903b7 to
f6c037a
Compare
|
lgtm I'll defer to @jparrill to tag |
|
/test verify |
|
/lgtm |
|
@bryan-cox: This pull request references Jira Issue OCPBUGS-18879, which is valid. 3 validation(s) were run on this bug
Requesting review from QA contact: DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/cherry-pick release-4.13 |
|
@bryan-cox: once the present PR merges, I will cherry-pick it on top of release-4.13 in a new PR and assign it to you. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/cherry-pick release-4.14 cherry-pick on 4.13 was a mistake. Will close that PR when it opens. |
|
@bryan-cox: once the present PR merges, I will cherry-pick it on top of release-4.14 in a new PR and assign it to you. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: bryan-cox, elsapassaro, enxebre The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
@bryan-cox: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
@bryan-cox: Jira Issue OCPBUGS-18879: All pull requests linked via external trackers have merged: Jira Issue OCPBUGS-18879 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@bryan-cox: #2950 failed to apply on top of branch "release-4.13": DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@bryan-cox: new pull request created: #3064 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
Fix included in accepted release 4.15.0-0.nightly-2023-09-30-003019 |
What this PR does / why we need it:
This PR contains the work needed to support HCP in a disconnected environment. The majority of the work in this PR is to determine image metadata in a disconnected environment, include user-ca bundles in appropriate HCP pods, and fixing the ignition server to support extracting image files in disconnected environments.
OCPBUGS-11939 - Get the image metadata from any ICSP/IDMS in a management cluster allowing a hosted cluster to be created in a disconnected environment with image mirror registry data from the ICSP/IDMS.
OCPBUGS-18128 - Automatically include the user created CA ConfigMap in the CPO deployment. This is accomplished in the HCC by combining the image registry CAs, listed in an Image (image.config.openshift.io) resource named 'clusters' in the 'openshift-config' namespace, to the CAs in the openshift-config-managed-trusted-ca-bundle ConfigMap from the
'hypershift' namespace and saving this to a new ConfigMap in the hostedcluster namespace. This new ConfigMap is then mounted as a volume in the CPO deployment.
OCPBUGS-18460 - Includes the user created CA ConfigMap in the ignition server deployment. This is accomplished by checking for the existence of a ConfigMap containing the needed CA data and mounting it to the ignition server deployment if the ConfigMap exists.
OCPBUGS-18602 - Includes the OpenShiftTrustedCABundle ConfigMap in the HCCO deployment.
OCPBUGS-18602 - Updates the HCCO release provider to include registry overrides and OpenShift image registry overrides.
OCPBUGS-18879 - Adds functionality to support extracting image files into release-manifests/image-reference config directory when working in a disconnected environment in the ignition server. Also, now catches the error when extracting image references to the config directory.
Which issue(s) this PR fixes:
Checklist