STOR-1445: Sync 10_deployment-hypershift.yaml from cluster-storage-operator repo - #2936
STOR-1445: Sync 10_deployment-hypershift.yaml from cluster-storage-operator repo#2936mpatlasov wants to merge 1 commit into
10_deployment-hypershift.yaml from cluster-storage-operator repo#2936Conversation
…-operator` repo The operator mounts secrets (with `tls.key` and `tls.cert`) and need to be restarted if they change.
|
@mpatlasov: This pull request references STOR-1445 which is a valid jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
1 similar comment
|
@mpatlasov: This pull request references STOR-1445 which is a valid jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
@mpatlasov: GitHub didn't allow me to request PR reviews from the following users: openshift/storage. Note that only openshift members and repo collaborators can review this PR, and authors cannot review their own PRs. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: mpatlasov The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/test e2e-kubevirt-aws-ovn |
|
@mpatlasov: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
/lgtm |
| - args: | ||
| - start | ||
| - -v=2 | ||
| - --terminate-on-files=/var/run/secrets/serving-cert/tls.crt |
There was a problem hiding this comment.
I don't see where this is mounted into the pod
There was a problem hiding this comment.
Thanks for nice catch, @csrwng ! It is mounted in standalone version of deployment (https://github.com/openshift/cluster-storage-operator/blob/master/manifests/10_deployment.yaml#L151) but not in hypershift. I'll close this PR as erroneous.
|
The path |
What this PR does / why we need it:
The PR has a commit:
which restarts CSO if the secret
cluster-storage-operator-serving-certis updated. This PR simply syncs those changes to hypershift repo asset yaml./cc @openshift/storage
Checklist