Skip to content

Rename target cluster kubeconfig - #23

Merged
openshift-merge-robot merged 1 commit into
openshift:mainfrom
enxebre:kubeconfig
Feb 12, 2021
Merged

Rename target cluster kubeconfig#23
openshift-merge-robot merged 1 commit into
openshift:mainfrom
enxebre:kubeconfig

Conversation

@enxebre

@enxebre enxebre commented Feb 12, 2021

Copy link
Copy Markdown
Member

Rename to a known predictable name so it can be tooling like autoscaling can be coonfigured to consume it ahead of time

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: enxebre

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 12, 2021
@enxebre enxebre mentioned this pull request Feb 12, 2021
@ironcladlou

Copy link
Copy Markdown
Contributor

I think you missed a spot in e2e. This object really should be wrapped up behind a function to make this kind of change easier

Rename to a known predictable name so it can be tooling like autoscaling can be coonfigured to consume it ahead of time
@enxebre

enxebre commented Feb 12, 2021

Copy link
Copy Markdown
Member Author

I think you missed a spot in e2e. This object really should be wrapped up behind a function to make this kind of change easier

Fixed!

@ironcladlou

Copy link
Copy Markdown
Contributor

/lgtm

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants