Skip to content

Rename control-plane-operator to hosted-cluster-config-operator - #12

Merged
openshift-merge-robot merged 1 commit into
openshift:mainfrom
ironcladlou:rename-control-plane-operator
Feb 3, 2021
Merged

Rename control-plane-operator to hosted-cluster-config-operator#12
openshift-merge-robot merged 1 commit into
openshift:mainfrom
ironcladlou:rename-control-plane-operator

Conversation

@ironcladlou

Copy link
Copy Markdown
Contributor

The control-plane-operator is responsible only for managing certain parts
of the configuration of a hosted control plane rather than the control plane
itself as a whole. Rename the operator to reflect its responsibilities and
allow the control-plane-operator name to be reclaimed by a new component
which will assume those broader responsibilities.

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ironcladlou

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 3, 2021
@ironcladlou
ironcladlou force-pushed the rename-control-plane-operator branch 3 times, most recently from c14e34d to cdaa350 Compare February 3, 2021 15:50
@ironcladlou

Copy link
Copy Markdown
Contributor Author
/Library/Developer/CommandLineTools/usr/bin/make -C test/e2e run
mkdir -p /Users/dmace/Projects/hypershift/test/e2e/data
KUSTOMIZE_PLUGIN_HOME=/Users/dmace/Projects/hypershift/config/example-cluster/plugin kustomize build --enable_alpha_plugins /Users/dmace/Projects/hypershift/config/example-cluster > /Users/dmace/Projects/hypershift/test/e2e/data/example-cluster.yaml
/Users/dmace/Projects/hypershift/hack/tools/bin/ginkgo -v -trace -tags=e2e -focus="" -nodes=1 --noColor=false  . -- \
          -e2e.data-dir="/Users/dmace/Projects/hypershift/test/e2e/data"
Running Suite: hypershift-e2e
=============================
Random Seed: 1612367367
Will run 1 of 1 specs

I0203 10:49:33.288383   62035 request.go:645] Throttling request took 1.015257692s, request: GET:https://api.dmace-c0d3.devcluster.openshift.com:6443/apis/helm.openshift.io/v1beta1?timeout=32s
When following the HyperShift quick-start [PR-Blocking] 
  Should create a functional guest cluster
  /Users/dmace/Projects/hypershift/test/e2e/quick_start_test.go:65
STEP: Applying the example cluster resources
INFO: Waiting for cluster resource to exist
STEP: Ensuring the guest cluster exposes a valid kubeconfig
INFO: Waiting for guest kubeconfig to become available
STEP: Establishing a connection to the guest apiserver
STEP: Ensuring guest nodes become ready
INFO: found 1 nodes
STEP: Deleting the example cluster
STEP: Ensuring the example cluster resources are deleted
INFO: Waiting for guest cluster namespace to be deleted
INFO: Waiting for the cluster resource to be deleted

• [SLOW TEST:509.694 seconds]
When following the HyperShift quick-start [PR-Blocking]
/Users/dmace/Projects/hypershift/test/e2e/quick_start_test.go:28
  Should create a functional guest cluster
  /Users/dmace/Projects/hypershift/test/e2e/quick_start_test.go:65
------------------------------

Ran 1 of 1 Specs in 512.682 seconds
SUCCESS! -- 1 Passed | 0 Failed | 0 Pending | 0 Skipped
PASS

Ginkgo ran 1 suite in 8m37.252963695s
Test Suite Passed

@ironcladlou
ironcladlou force-pushed the rename-control-plane-operator branch from cdaa350 to 45861ed Compare February 3, 2021 17:14
The control-plane-operator is responsible only for managing certain parts
of the configuration of a hosted control plane rather than the control plane
itself as a whole. Rename the operator to reflect its responsibilities and
allow the control-plane-operator name to be reclaimed by a new component
which will assume those broader responsibilities.
@ironcladlou
ironcladlou force-pushed the rename-control-plane-operator branch from 45861ed to 50ec715 Compare February 3, 2021 17:15
@csrwng

csrwng commented Feb 3, 2021

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Feb 3, 2021
@openshift-merge-robot
openshift-merge-robot merged commit a207c61 into openshift:main Feb 3, 2021
zanetworker pushed a commit to zanetworker/hypershift that referenced this pull request Sep 28, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants