Skip to content

OCPBUGS-38522: Fix the destination of asset's copy - #1112

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
jmencak:4.17-1058-localbin-fix
Jul 22, 2024
Merged

OCPBUGS-38522: Fix the destination of asset's copy#1112
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
jmencak:4.17-1058-localbin-fix

Conversation

@jmencak

@jmencak jmencak commented Jul 16, 2024

Copy link
Copy Markdown
Contributor

#1058 introduced an issue, where assets meant to be copied to /usr/local/bin were being copied to /usr/local/bin/bin. Fix this. Additional cleanup (duplicate statements in if/else) to follow.

/cc @jupierce

PR1058 introduced an issue, where assets meant to be copied to
/usr/local/bin were being copied to /usr/local/bin/bin.  Fix this.
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jul 16, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jmencak: This pull request explicitly references no jira issue.

Details

In response to this:

#1058 introduced an issue, where assets meant to be copied to /usr/local/bin were being copied to /usr/local/bin/bin. Fix this. Additional cleanup (duplicate statements in if/else) to follow.

/cc @jupierce

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested a review from jupierce July 16, 2024 12:15
@openshift-ci

openshift-ci Bot commented Jul 16, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jmencak

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 16, 2024
@jupierce

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 18, 2024
@jmencak

jmencak commented Jul 22, 2024

Copy link
Copy Markdown
Contributor Author

/test all

@jmencak

jmencak commented Jul 22, 2024

Copy link
Copy Markdown
Contributor Author

/label acknowledge-critical-fixes-only

@openshift-ci openshift-ci Bot added the acknowledge-critical-fixes-only Indicates if the issuer of the label is OK with the policy. label Jul 22, 2024
@openshift-ci

openshift-ci Bot commented Jul 22, 2024

Copy link
Copy Markdown
Contributor

@jmencak: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 10bb4c6 into openshift:master Jul 22, 2024
@jmencak
jmencak deleted the 4.17-1058-localbin-fix branch July 22, 2024 18:29
@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

Distgit: cluster-node-tuning-operator
This PR has been included in build cluster-node-tuning-operator-container-v4.17.0-202407222110.p0.g10bb4c6.assembly.stream.el9.
All builds following this will include this PR.

@jmencak jmencak changed the title NO-JIRA: Fix the destination of asset's copy OCPBUGS-38522: Fix the destination of asset's copy Aug 15, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jmencak: Jira Issue OCPBUGS-38522: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-38522 has been moved to the MODIFIED state.

Details

In response to this:

#1058 introduced an issue, where assets meant to be copied to /usr/local/bin were being copied to /usr/local/bin/bin. Fix this. Additional cleanup (duplicate statements in if/else) to follow.

/cc @jupierce

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@jmencak

jmencak commented Aug 15, 2024

Copy link
Copy Markdown
Contributor Author

/jira refresh

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@jmencak: Jira Issue OCPBUGS-38522 is in an unrecognized state (MODIFIED) and will not be moved to the MODIFIED state.

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@jmencak

jmencak commented Aug 15, 2024

Copy link
Copy Markdown
Contributor Author

/cherry-pick release-4.16

@openshift-cherrypick-robot

Copy link
Copy Markdown

@jmencak: new pull request created: #1134

Details

In response to this:

/cherry-pick release-4.16

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

acknowledge-critical-fixes-only Indicates if the issuer of the label is OK with the policy. approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants