Skip to content

[release-4.14] OCPBUGS-22787,OCPBUGS-22788,OCPBUGS-22789: ovnkube: container scripts cleanup - #2090

Merged
openshift-merge-bot[bot] merged 24 commits into
openshift:release-4.14from
dcbw:414-db-script-cleanup
Dec 13, 2023
Merged

[release-4.14] OCPBUGS-22787,OCPBUGS-22788,OCPBUGS-22789: ovnkube: container scripts cleanup#2090
openshift-merge-bot[bot] merged 24 commits into
openshift:release-4.14from
dcbw:414-db-script-cleanup

Conversation

@dcbw

@dcbw dcbw commented Nov 1, 2023

Copy link
Copy Markdown
Contributor

Put common scripts into a bash script library and source that library from the ovnkube-node containers.

@openshift-ci
openshift-ci Bot requested review from abhat and dougbtv November 1, 2023 13:53
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 1, 2023
@dcbw dcbw changed the title [release-4.14] ovnkube: container scripts cleanup [release-4.14] OCPBUGS-22787,OCPBUGS-22788,OCPBUGS-22789: ovnkube: container scripts cleanup Nov 1, 2023
@openshift-ci-robot openshift-ci-robot added jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. labels Nov 1, 2023
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dcbw: This pull request references Jira Issue OCPBUGS-22787, which is invalid:

  • expected dependent Jira Issue OCPBUGS-18089 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ON_QA instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

This pull request references Jira Issue OCPBUGS-22788, which is invalid:

  • expected dependent Jira Issue OCPBUGS-18088 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ON_QA instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

This pull request references Jira Issue OCPBUGS-22789, which is invalid:

  • expected Jira Issue OCPBUGS-22789 to depend on a bug targeting a version in 4.15.0 and in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but no dependents were found

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Put common scripts into a bash script library and source that library from the ovnkube-node containers.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. label Nov 1, 2023
@dcbw

dcbw commented Nov 1, 2023

Copy link
Copy Markdown
Contributor Author

/jira refresh

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dcbw: This pull request references Jira Issue OCPBUGS-22787, which is invalid:

  • expected dependent Jira Issue OCPBUGS-18089 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ON_QA instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

This pull request references Jira Issue OCPBUGS-22788, which is invalid:

  • expected dependent Jira Issue OCPBUGS-18088 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ON_QA instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

This pull request references Jira Issue OCPBUGS-22789, which is valid. The bug has been moved to the POST state.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state New, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-15201 is in the state Closed (Done-Errata), which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-15201 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @zhaozhanqi

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci
openshift-ci Bot requested a review from zhaozhanqi November 1, 2023 13:57
@droslean

droslean commented Nov 1, 2023

Copy link
Copy Markdown
Member

/retest

@dcbw

dcbw commented Nov 1, 2023

Copy link
Copy Markdown
Contributor Author

/retest

openshift/release#45159

@dcbw

dcbw commented Nov 1, 2023

Copy link
Copy Markdown
Contributor Author

/retest

@dcbw

dcbw commented Nov 8, 2023

Copy link
Copy Markdown
Contributor Author

/jira refresh

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dcbw: This pull request references Jira Issue OCPBUGS-22787, which is valid. The bug has been moved to the POST state.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state New, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18089 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18089 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @anuragthehatter

This pull request references Jira Issue OCPBUGS-22788, which is invalid:

  • expected dependent Jira Issue OCPBUGS-18088 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ON_QA instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

This pull request references Jira Issue OCPBUGS-22789, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-15201 is in the state Closed (Done-Errata), which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-15201 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @zhaozhanqi

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@dcbw

dcbw commented Nov 14, 2023

Copy link
Copy Markdown
Contributor Author

/jira refresh

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dcbw: This pull request references Jira Issue OCPBUGS-22787, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18089 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18089 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @anuragthehatter

This pull request references Jira Issue OCPBUGS-22788, which is invalid:

  • expected dependent Jira Issue OCPBUGS-18088 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ON_QA instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

This pull request references Jira Issue OCPBUGS-22789, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-15201 is in the state Closed (Done-Errata), which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-15201 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @zhaozhanqi

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@dcbw

dcbw commented Nov 20, 2023

Copy link
Copy Markdown
Contributor Author

/test ci/prow/4.14-upgrade-from-stable-4.13-e2e-gcp-ovn-upgrade

@openshift-ci

openshift-ci Bot commented Nov 20, 2023

Copy link
Copy Markdown
Contributor

@dcbw: The specified target(s) for /test were not found.
The following commands are available to trigger required jobs:

  • /test 4.14-upgrade-from-stable-4.13-images
  • /test e2e-aws-ovn-network-migration
  • /test e2e-aws-ovn-windows
  • /test e2e-aws-sdn-multi
  • /test e2e-aws-sdn-network-migration-rollback
  • /test e2e-aws-sdn-network-reverse-migration
  • /test e2e-gcp-ovn
  • /test e2e-gcp-ovn-upgrade
  • /test e2e-gcp-sdn
  • /test e2e-hypershift-ovn
  • /test e2e-metal-ipi-ovn-ipv6
  • /test e2e-vsphere-ovn-windows
  • /test images
  • /test lint
  • /test unit
  • /test verify

The following commands are available to trigger optional jobs:

  • /test 4.14-upgrade-from-stable-4.13-e2e-aws-ovn-upgrade
  • /test 4.14-upgrade-from-stable-4.13-e2e-azure-ovn-upgrade
  • /test 4.14-upgrade-from-stable-4.13-e2e-gcp-ovn-upgrade
  • /test e2e-aws-hypershift-ovn-kubevirt
  • /test e2e-aws-ovn-local-to-shared-gateway-mode-migration
  • /test e2e-aws-ovn-serial
  • /test e2e-aws-ovn-shared-to-local-gateway-mode-migration-periodic
  • /test e2e-aws-ovn-single-node
  • /test e2e-aws-sdn-upgrade
  • /test e2e-azure-ovn
  • /test e2e-azure-ovn-dualstack
  • /test e2e-azure-ovn-manual-oidc
  • /test e2e-metal-ipi-ovn-ipv6-ipsec
  • /test e2e-network-mtu-migration-ovn-ipv4
  • /test e2e-network-mtu-migration-ovn-ipv6
  • /test e2e-network-mtu-migration-sdn-ipv4
  • /test e2e-openstack-kuryr
  • /test e2e-openstack-ovn
  • /test e2e-openstack-sdn
  • /test e2e-ovn-hybrid-step-registry
  • /test e2e-ovn-ipsec-step-registry
  • /test e2e-ovn-step-registry
  • /test e2e-vsphere-ovn
  • /test e2e-vsphere-ovn-dualstack
  • /test e2e-vsphere-ovn-dualstack-primaryv6
  • /test qe-perfscale-aws-ovn-medium-cluster-density
  • /test qe-perfscale-aws-ovn-medium-node-density-cni
  • /test qe-perfscale-aws-ovn-small-cluster-density
  • /test qe-perfscale-aws-ovn-small-node-density-cni

Use /test all to run the following jobs that were automatically triggered:

  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-e2e-aws-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-e2e-azure-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-e2e-gcp-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-images
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-hypershift-ovn-kubevirt
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-local-to-shared-gateway-mode-migration
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-network-migration
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-serial
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-shared-to-local-gateway-mode-migration-periodic
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-single-node
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-windows
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-multi
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-network-migration-rollback
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-network-reverse-migration
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-azure-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-azure-ovn-dualstack
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-gcp-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-gcp-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-gcp-sdn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-hypershift-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-metal-ipi-ovn-ipv6
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-metal-ipi-ovn-ipv6-ipsec
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-network-mtu-migration-ovn-ipv4
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-network-mtu-migration-ovn-ipv6
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-network-mtu-migration-sdn-ipv4
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-openstack-kuryr
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-openstack-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-openstack-sdn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-ovn-hybrid-step-registry
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-ovn-ipsec-step-registry
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-ovn-step-registry
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn-dualstack
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn-dualstack-primaryv6
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn-windows
  • pull-ci-openshift-cluster-network-operator-release-4.14-images
  • pull-ci-openshift-cluster-network-operator-release-4.14-lint
  • pull-ci-openshift-cluster-network-operator-release-4.14-unit
  • pull-ci-openshift-cluster-network-operator-release-4.14-verify
Details

In response to this:

/test ci/prow/4.14-upgrade-from-stable-4.13-e2e-gcp-ovn-upgrade

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@dcbw

dcbw commented Nov 20, 2023

Copy link
Copy Markdown
Contributor Author

/test 4.14-upgrade-from-stable-4.13-e2e-gcp-ovn-upgrade

flavio-fernandes and others added 8 commits November 21, 2023 15:59
During an internal audit conducted by one of our customers in OCP, it was
discovered that the openshift-sdn component, specifically on port 9101,
has exposed certain cipher suites that are considered weak. We have
identified the following weak ciphers that are currently being accepted:

TLS_ECDHE_ECDSA_WITH_AES_128_CBC_SHA256 - See [1]
TLS_ECDHE_RSA_WITH_AES_128_CBC_SHA256 - See [2]

[1] https://ciphersuite.info/cs/TLS_ECDHE_ECDSA_WITH_AES_128_CBC_SHA256/
[2] https://ciphersuite.info/cs/TLS_ECDHE_RSA_WITH_AES_128_CBC_SHA256/

Reported-at: https://issues.redhat.com/browse/OCPBUGS-15201
Signed-off-by: Flavio Fernandes <flaviof@redhat.com>
Like GENEVE we don't need to conntrack the hybrid overlay VXLAN
traffic.

Resolves: https://issues.redhat.com/browse/OCPBUGS-10652

Signed-off-by: Dan Williams <dcbw@redhat.com>
ovn-ctl didn't originally handle upgrading clustered databases,
but Numan added that to CNO via
openshift#755

He then added clustered DB upgrade support to ovn-ctl itself
in September 2020:

67e2f386cc838d0b0f9b4b5da7fe611e1113b70c
"ovn-ctl: Handle cluster db upgrades for run_(nb/sb)_ovsdb"
https://patchwork.ozlabs.org/project/ovn/patch/20200909071939.5095-1-numans@ovn.org/

And we just never removed the upgrade bits from the CNO
script. But also, standalone DB upgrades happen before
the ovsdb-server starts so the original problem doesn't
even apply anymore.
We don't need to wait for RAFT clusters to form anymore so we
don't have to wait quite as long for readiness. We might have to
wait a bit for a DB conversion, but that's pretty fast these days.
No reason it needs to be in the readiness probe and executed over
and over. Put it along with the other options that get set after
the DB gets started, like probe intervals and NB_Globals.
Since there's no RAFT, there's no cluster to converge and we don't
need to wait.
@openshift-ci-robot openshift-ci-robot removed the jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. label Dec 12, 2023
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dcbw: This pull request references Jira Issue OCPBUGS-22787, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18089 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18089 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @anuragthehatter

This pull request references Jira Issue OCPBUGS-22788, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-18088 is in the state Verified, which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-18088 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @anuragthehatter

This pull request references Jira Issue OCPBUGS-22789, which is valid.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.14.z) matches configured target version for branch (4.14.z)
  • bug is in the state POST, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-15201 is in the state Closed (Done-Errata), which is one of the valid states (VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA))
  • dependent Jira Issue OCPBUGS-15201 targets the "4.15.0" version, which is one of the valid target versions: 4.15.0
  • bug has dependents

Requesting review from QA contact:
/cc @zhaozhanqi

Details

In response to this:

/jira refresh

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@dcbw dcbw added bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. labels Dec 12, 2023
@asood-rh

Copy link
Copy Markdown
Contributor

/label cherry-pick-approved

@openshift-ci openshift-ci Bot added the cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. label Dec 12, 2023
@jluhrsen

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Dec 12, 2023
@openshift-ci

openshift-ci Bot commented Dec 12, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dcbw, jluhrsen

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@jluhrsen

Copy link
Copy Markdown
Contributor

/test ci/prow/e2e-vsphere-ovn-windows

@openshift-ci

openshift-ci Bot commented Dec 13, 2023

Copy link
Copy Markdown
Contributor

@jluhrsen: The specified target(s) for /test were not found.
The following commands are available to trigger required jobs:

  • /test 4.14-upgrade-from-stable-4.13-images
  • /test e2e-aws-ovn-network-migration
  • /test e2e-aws-ovn-windows
  • /test e2e-aws-sdn-multi
  • /test e2e-aws-sdn-network-migration-rollback
  • /test e2e-aws-sdn-network-reverse-migration
  • /test e2e-gcp-ovn
  • /test e2e-gcp-ovn-upgrade
  • /test e2e-gcp-sdn
  • /test e2e-hypershift-ovn
  • /test e2e-metal-ipi-ovn-ipv6
  • /test e2e-vsphere-ovn-windows
  • /test images
  • /test lint
  • /test unit
  • /test verify

The following commands are available to trigger optional jobs:

  • /test 4.14-upgrade-from-stable-4.13-e2e-aws-ovn-upgrade
  • /test 4.14-upgrade-from-stable-4.13-e2e-azure-ovn-upgrade
  • /test 4.14-upgrade-from-stable-4.13-e2e-gcp-ovn-upgrade
  • /test e2e-aws-hypershift-ovn-kubevirt
  • /test e2e-aws-ovn-local-to-shared-gateway-mode-migration
  • /test e2e-aws-ovn-serial
  • /test e2e-aws-ovn-shared-to-local-gateway-mode-migration-periodic
  • /test e2e-aws-ovn-single-node
  • /test e2e-aws-sdn-upgrade
  • /test e2e-azure-ovn
  • /test e2e-azure-ovn-dualstack
  • /test e2e-azure-ovn-manual-oidc
  • /test e2e-metal-ipi-ovn-ipv6-ipsec
  • /test e2e-network-mtu-migration-ovn-ipv4
  • /test e2e-network-mtu-migration-ovn-ipv6
  • /test e2e-network-mtu-migration-sdn-ipv4
  • /test e2e-openstack-kuryr
  • /test e2e-openstack-ovn
  • /test e2e-openstack-sdn
  • /test e2e-ovn-hybrid-step-registry
  • /test e2e-ovn-ipsec-step-registry
  • /test e2e-ovn-step-registry
  • /test e2e-vsphere-ovn
  • /test e2e-vsphere-ovn-dualstack
  • /test e2e-vsphere-ovn-dualstack-primaryv6
  • /test qe-perfscale-aws-ovn-medium-cluster-density
  • /test qe-perfscale-aws-ovn-medium-node-density-cni
  • /test qe-perfscale-aws-ovn-small-cluster-density
  • /test qe-perfscale-aws-ovn-small-node-density-cni

Use /test all to run the following jobs that were automatically triggered:

  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-e2e-aws-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-e2e-azure-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-e2e-gcp-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-4.14-upgrade-from-stable-4.13-images
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-hypershift-ovn-kubevirt
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-local-to-shared-gateway-mode-migration
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-network-migration
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-serial
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-shared-to-local-gateway-mode-migration-periodic
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-single-node
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-ovn-windows
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-multi
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-network-migration-rollback
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-network-reverse-migration
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-aws-sdn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-azure-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-azure-ovn-dualstack
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-gcp-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-gcp-ovn-upgrade
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-gcp-sdn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-hypershift-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-metal-ipi-ovn-ipv6
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-metal-ipi-ovn-ipv6-ipsec
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-network-mtu-migration-ovn-ipv4
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-network-mtu-migration-ovn-ipv6
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-network-mtu-migration-sdn-ipv4
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-openstack-kuryr
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-openstack-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-openstack-sdn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-ovn-hybrid-step-registry
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-ovn-ipsec-step-registry
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-ovn-step-registry
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn-dualstack
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn-dualstack-primaryv6
  • pull-ci-openshift-cluster-network-operator-release-4.14-e2e-vsphere-ovn-windows
  • pull-ci-openshift-cluster-network-operator-release-4.14-images
  • pull-ci-openshift-cluster-network-operator-release-4.14-lint
  • pull-ci-openshift-cluster-network-operator-release-4.14-unit
  • pull-ci-openshift-cluster-network-operator-release-4.14-verify
Details

In response to this:

/test ci/prow/e2e-vsphere-ovn-windows

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@jluhrsen

Copy link
Copy Markdown
Contributor

/test e2e-vsphere-ovn-windows

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD 34cf2de and 2 for PR HEAD 0d21e57 in total

@openshift-ci

openshift-ci Bot commented Dec 13, 2023

Copy link
Copy Markdown
Contributor

@dcbw: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-vsphere-ovn-dualstack-primaryv6 0d21e57 link false /test e2e-vsphere-ovn-dualstack-primaryv6
ci/prow/e2e-azure-ovn-dualstack 0d21e57 link false /test e2e-azure-ovn-dualstack
ci/prow/e2e-openstack-kuryr 0d21e57 link false /test e2e-openstack-kuryr
ci/prow/e2e-aws-hypershift-ovn-kubevirt 0d21e57 link false /test e2e-aws-hypershift-ovn-kubevirt

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@dcbw

dcbw commented Dec 13, 2023

Copy link
Copy Markdown
Contributor Author

/test e2e-aws-sdn-network-reverse-migration

@dcbw

dcbw commented Dec 13, 2023

Copy link
Copy Markdown
Contributor Author

/tide refresh

@openshift-merge-bot
openshift-merge-bot Bot merged commit 0ede33d into openshift:release-4.14 Dec 13, 2023
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@dcbw: Jira Issue OCPBUGS-22787: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-22787 has been moved to the MODIFIED state.

Jira Issue OCPBUGS-22788: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-22788 has been moved to the MODIFIED state.

Jira Issue OCPBUGS-22789: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-22789 has been moved to the MODIFIED state.

Details

In response to this:

Put common scripts into a bash script library and source that library from the ovnkube-node containers.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-bot

Copy link
Copy Markdown
Contributor

[ART PR BUILD NOTIFIER]

This PR has been included in build cluster-network-operator-container-v4.14.0-202312132033.p0.g0ede33d.assembly.stream for distgit cluster-network-operator.
All builds following this will include this PR.

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

Fix included in accepted release 4.14.0-0.nightly-2023-12-14-000637

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.