Skip to content

[release-4.12] OCPBUGS-4040: Restart authentication operator if console capability is enabled - #593

Merged
openshift-merge-robot merged 1 commit into
openshift:release-4.12from
openshift-cherrypick-robot:cherry-pick-589-to-release-4.12
Nov 29, 2022
Merged

[release-4.12] OCPBUGS-4040: Restart authentication operator if console capability is enabled#593
openshift-merge-robot merged 1 commit into
openshift:release-4.12from
openshift-cherrypick-robot:cherry-pick-589-to-release-4.12

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #589

/assign bparees

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: Jira Issue OCPBUGS-3440 has been cloned as Jira Issue OCPBUGS-4040. Retitling PR to link against new bug.
/retitle [release-4.12] OCPBUGS-4040: Restart authentication operator if console capability is enabled

Details

In response to this:

This is an automated cherry-pick of #589

/assign bparees

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot changed the title [release-4.12] OCPBUGS-3440: Restart authentication operator if console capability is enabled [release-4.12] OCPBUGS-4040: Restart authentication operator if console capability is enabled Nov 23, 2022
@openshift-ci-robot openshift-ci-robot added jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. labels Nov 23, 2022
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: This pull request references Jira Issue OCPBUGS-4040, which is valid. The bug has been moved to the POST state.

6 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target version (4.12.0) matches configured target version for branch (4.12.0)
  • bug is in the state New, which is one of the valid states (NEW, ASSIGNED, POST)
  • dependent bug Jira Issue OCPBUGS-3440 is in the state MODIFIED, which is one of the valid states (MODIFIED, ON_QA, VERIFIED)
  • dependent Jira Issue OCPBUGS-3440 targets the "4.13.0" version, which is one of the valid target versions: 4.13.0
  • bug has dependents

Requesting review from QA contact:
/cc @xingxingxia

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

This is an automated cherry-pick of #589

/assign bparees

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci
openshift-ci Bot requested a review from xingxingxia November 23, 2022 14:52
@openshift-ci

openshift-ci Bot commented Nov 23, 2022

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: No Bugzilla bug is referenced in the title of this pull request.
To reference a bug, add 'Bug XXX:' to the title of this pull request and request another bug refresh with /bugzilla refresh.

Retaining the bugzilla/valid-bug label as it was manually added.

Details

In response to this:

[release-4.12] OCPBUGS-4040: Restart authentication operator if console capability is enabled

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci
openshift-ci Bot requested review from stlaz and sttts November 23, 2022 14:52
@bparees

bparees commented Nov 23, 2022

Copy link
Copy Markdown
Contributor

@stlaz 4.12 backport, ptal.

@stlaz

stlaz commented Nov 25, 2022

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Nov 25, 2022
@stlaz

stlaz commented Nov 25, 2022

Copy link
Copy Markdown
Contributor

/label backport-risk-assessed

@openshift-ci

openshift-ci Bot commented Nov 25, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: openshift-cherrypick-robot, stlaz

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Nov 25, 2022
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

/retest-required

Remaining retests: 0 against base HEAD d064e84 and 2 for PR HEAD 8353151 in total

@xingxingxia

Copy link
Copy Markdown
Contributor

/label cherry-pick-approved

@openshift-ci openshift-ci Bot added the cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. label Nov 28, 2022
@bparees

bparees commented Nov 28, 2022

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-ci

openshift-ci Bot commented Nov 28, 2022

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-single-node 8353151 link false /test e2e-aws-single-node
ci/prow/e2e-agnostic-ipv6 8353151 link false /test e2e-agnostic-ipv6

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@bparees

bparees commented Nov 29, 2022

Copy link
Copy Markdown
Contributor

/retest-required

@openshift-merge-robot
openshift-merge-robot merged commit 54844a4 into openshift:release-4.12 Nov 29, 2022
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: All pull requests linked via external trackers have merged:

Jira Issue OCPBUGS-4040 has been moved to the MODIFIED state.

Details

In response to this:

This is an automated cherry-pick of #589

/assign bparees

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. cherry-pick-approved Indicates a cherry-pick PR into a release branch has been approved by the release branch manager. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants