Skip to content

pj-rehearse: Disable rehearsals on tests where the 'restrict_network_access' config is toggled to 'false' - #4348

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
smg247:disabled-rehearse-on-network-toggle-to-false
Sep 27, 2024
Merged

pj-rehearse: Disable rehearsals on tests where the 'restrict_network_access' config is toggled to 'false'#4348
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
smg247:disabled-rehearse-on-network-toggle-to-false

Conversation

@smg247

@smg247 smg247 commented Sep 26, 2024

Copy link
Copy Markdown
Member

This is done to disallow rehearsing any test that bypasses the firewall without first having someone reviewed the configuration.

Iff we find other cases where specific fields being changed should disable the entire rehearsal, we could make this logic more general. For now, this is sufficient.

For: https://issues.redhat.com/browse/DPTP-4171

…g has been set to 'false' on the candidate ref.

This is done to disallow rehearsing any test that bypasses the firewall without first having someone reviewed the configuration.
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 26, 2024
@smg247

smg247 commented Sep 26, 2024

Copy link
Copy Markdown
Member Author

/hold so I can monitor the rollout

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 26, 2024
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 26, 2024
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Sep 26, 2024
@smg247

smg247 commented Sep 26, 2024

Copy link
Copy Markdown
Member Author

/label tide/merge-method-squash

@openshift-ci openshift-ci Bot added the tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. label Sep 26, 2024
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

/test remaining-required

@smg247

smg247 commented Sep 27, 2024

Copy link
Copy Markdown
Member Author

/retest-required

@openshift-ci

openshift-ci Bot commented Sep 27, 2024

Copy link
Copy Markdown
Contributor

@smg247: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/security a0e889f link false /test security

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 27, 2024
@openshift-ci

openshift-ci Bot commented Sep 27, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bear-redhat, smg247

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@smg247

smg247 commented Sep 27, 2024

Copy link
Copy Markdown
Member Author

/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 27, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit a92c8bb into openshift:master Sep 27, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants