Skip to content

NO-ISSUE: Remove failing NMState kube api test until it is fixed - #6839

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
danmanor:remove-failing-nmstate-kubeapi-test
Sep 29, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:masterfrom
danmanor:remove-failing-nmstate-kubeapi-test

Conversation

@danmanor

@danmanor danmanor commented Sep 29, 2024 •

Copy link
Copy Markdown
Contributor

Due to a temporary issue, the kube-api subsystem tests passed falsely. Thus, masking the failure of deploy clusterDeployment and infraEnv and with an invalid NMState config YAML in #6695. See job. To avoid blocking PRs, I am temporarily removing this test until it is fixed.

A follow up issue - https://issues.redhat.com/browse/MGMT-19027

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 29, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@danmanor: This pull request explicitly references no jira issue.

Details

In response to this:

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 29, 2024
@openshift-ci
openshift-ci Bot requested review from eifrach and rccrdpccl September 29, 2024 12:32
@openshift-ci

openshift-ci Bot commented Sep 29, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: danmanor

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 29, 2024
@danmanor

Copy link
Copy Markdown
Contributor Author

/override ci/prow/edge-subsystem-kubeapi-aws ci/prow/edge-subsystem-aws ci/prow/e2e-agent-compact-ipv4

@openshift-ci

openshift-ci Bot commented Sep 29, 2024

Copy link
Copy Markdown
Contributor

@danmanor: Overrode contexts on behalf of danmanor: ci/prow/e2e-agent-compact-ipv4, ci/prow/edge-subsystem-aws, ci/prow/edge-subsystem-kubeapi-aws

Details

In response to this:

/override ci/prow/edge-subsystem-kubeapi-aws ci/prow/edge-subsystem-aws ci/prow/e2e-agent-compact-ipv4

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@danmanor

Copy link
Copy Markdown
Contributor Author

/cc @linoyaslan @danielerez

@danmanor

Copy link
Copy Markdown
Contributor Author

/retest

@linoyaslan

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 29, 2024
@codecov

codecov Bot commented Sep 29, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 68.73%. Comparing base (0fde766) to head (feb60da).
Report is 2 commits behind head on master.

Additional details and impacted files

Impacted file tree graph

@@           Coverage Diff           @@
##           master    #6839   +/-   ##
=======================================
  Coverage   68.73%   68.73%           
=======================================
  Files         249      249           
  Lines       37278    37278           
=======================================
  Hits        25624    25624           
  Misses       9373     9373           
  Partials     2281     2281           
Flag Coverage Δ
68.73% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

@danmanor

Copy link
Copy Markdown
Contributor Author

/override ci/prow/edge-subsystem-kubeapi-aws ci/prow/edge-subsystem-aws ci/prow/e2e-agent-compact-ipv4

@openshift-ci

openshift-ci Bot commented Sep 29, 2024

Copy link
Copy Markdown
Contributor

@danmanor: Overrode contexts on behalf of danmanor: ci/prow/e2e-agent-compact-ipv4, ci/prow/edge-subsystem-aws, ci/prow/edge-subsystem-kubeapi-aws

Details

In response to this:

/override ci/prow/edge-subsystem-kubeapi-aws ci/prow/edge-subsystem-aws ci/prow/e2e-agent-compact-ipv4

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 0c2f655 into openshift:master Sep 29, 2024
@openshift-ci

openshift-ci Bot commented Sep 29, 2024

Copy link
Copy Markdown
Contributor

@danmanor: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants