Skip to content

[release-ocm-2.9] MGMT-18585: Fix AutomatedCleaningMode behavior - #6670

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:release-ocm-2.9from
CrystalChun:fix-2.9
Aug 12, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:release-ocm-2.9from
CrystalChun:fix-2.9

Conversation

@CrystalChun

Copy link
Copy Markdown
Contributor

Manual cherry-pick/backport of #6662


AutomatedCleaningMode should not be used when the
converged flow is disabled since it requires IPA
which is only enabled when the converged flow is enabled.

Caused by regression in PR
#5319 which relies on users to set the automatedCleaningMode spec, but did not take into account the converged flow being disabled.

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

/cc @gamli75

AutomatedCleaningMode should not be used when the
converged flow is disabled since it requires IPA
which is only enabled when the converged flow is enabled.

Caused by regression in PR
openshift#5319
which relies on users to set the automatedCleaningMode
spec, but did not take into account the converged flow
being disabled.
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Aug 12, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 12, 2024 •

Copy link
Copy Markdown

@CrystalChun: This pull request references MGMT-18585 which is a valid jira issue.

Details

In response to this:

Manual cherry-pick/backport of #6662


AutomatedCleaningMode should not be used when the
converged flow is disabled since it requires IPA
which is only enabled when the converged flow is enabled.

Caused by regression in PR
#5319 which relies on users to set the automatedCleaningMode spec, but did not take into account the converged flow being disabled.

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

/cc @gamli75

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested a review from gamli75 August 12, 2024 20:20
@openshift-ci openshift-ci Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Aug 12, 2024
@codecov

codecov Bot commented Aug 12, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 67.70%. Comparing base (e8ff318) to head (2a121a5).

Additional details and impacted files

Impacted file tree graph

@@                 Coverage Diff                 @@
##           release-ocm-2.9    #6670      +/-   ##
===================================================
+ Coverage            67.68%   67.70%   +0.01%     
===================================================
  Files                  232      232              
  Lines                34119    34122       +3     
===================================================
+ Hits                 23095    23101       +6     
+ Misses                8976     8975       -1     
+ Partials              2048     2046       -2     
Files Coverage Δ
...nal/controller/controllers/bmh_agent_controller.go 75.86% <100.00%> (+0.08%) ⬆️

... and 2 files with indirect coverage changes

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 12, 2024
@openshift-ci

openshift-ci Bot commented Aug 12, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: CrystalChun, gamli75

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 12, 2024
@openshift-ci

openshift-ci Bot commented Aug 12, 2024

Copy link
Copy Markdown
Contributor

@CrystalChun: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 277a96e into openshift:release-ocm-2.9 Aug 12, 2024
@CrystalChun
CrystalChun deleted the fix-2.9 branch October 18, 2024 15:03
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants