Repository navigation
[release-ocm-2.10] MGMT-18584: Fix AutomatedCleaningMode behavior - #6668
Conversation
AutomatedCleaningMode should not be used when the converged flow is disabled since it requires IPA which is only enabled when the converged flow is enabled. Caused by regression in PR openshift#5319 which relies on users to set the automatedCleaningMode spec, but did not take into account the converged flow being disabled.
|
@openshift-cherrypick-robot: Ignoring requests to cherry-pick non-bug issues: MGMT-18560 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/retitle [release-ocm-2.10] MGMT-18584: Fix AutomatedCleaningMode behavior |
|
@openshift-cherrypick-robot: This pull request references MGMT-18584 which is a valid jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
/uncc @oourfali @romfreiman |
Codecov ReportAll modified and coverable lines are covered by tests ✅
Additional details and impacted files@@ Coverage Diff @@
## release-ocm-2.10 #6668 +/- ##
====================================================
+ Coverage 68.29% 68.30% +0.01%
====================================================
Files 236 236
Lines 35093 35096 +3
====================================================
+ Hits 23968 23974 +6
+ Misses 9051 9049 -2
+ Partials 2074 2073 -1
|
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: gamli75, openshift-cherrypick-robot The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
@openshift-cherrypick-robot: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
@openshift-cherrypick-robot: #6668 failed to apply on top of branch "release-ocm-2.9": DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
This is an automated cherry-pick of #6662
/assign CrystalChun
/cherrypick release-ocm-2.9