Skip to content

[release-ocm-2.8] MGMT-18313: Replace golang base image as it is based on Centos Linux 7 - #6657

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:release-ocm-2.8from
danmanor:replace-base-image-2.8
Aug 7, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:release-ocm-2.8from
danmanor:replace-base-image-2.8

Conversation

@danmanor

@danmanor danmanor commented Aug 6, 2024

Copy link
Copy Markdown
Contributor

Backport of #6637

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Aug 6, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 6, 2024 •

Copy link
Copy Markdown

@danmanor: This pull request references MGMT-18313 which is a valid jira issue.

Details

In response to this:

Backport of #6637

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Aug 6, 2024
@openshift-ci
openshift-ci Bot requested review from adriengentil and eifrach August 6, 2024 12:48
@openshift-ci openshift-ci Bot added the kind/dependency-change Categorizes issue or PR as related to changing dependencies label Aug 6, 2024
@openshift-ci

openshift-ci Bot commented Aug 6, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: danmanor

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 6, 2024
@codecov

codecov Bot commented Aug 6, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 67.32%. Comparing base (5e2272a) to head (a56b5c0).

Additional details and impacted files

Impacted file tree graph

@@                 Coverage Diff                 @@
##           release-ocm-2.8    #6657      +/-   ##
===================================================
- Coverage            67.33%   67.32%   -0.02%     
===================================================
  Files                  222      222              
  Lines                32803    32803              
===================================================
- Hits                 22089    22084       -5     
- Misses                8721     8725       +4     
- Partials              1993     1994       +1     

see 3 files with indirect coverage changes

@danmanor
danmanor force-pushed the replace-base-image-2.8 branch from 4708207 to a56b5c0 Compare August 7, 2024 07:10
@openshift-ci

openshift-ci Bot commented Aug 7, 2024

Copy link
Copy Markdown
Contributor

@danmanor: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@eifrach

eifrach commented Aug 7, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 7, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit bef18b3 into openshift:release-ocm-2.8 Aug 7, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. kind/dependency-change Categorizes issue or PR as related to changing dependencies lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants