Skip to content

[release-ocm-2.11] MGMT-18313: Replace golang base image as it is based on Centos Linux 7 - #6647

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:release-ocm-2.11from
danmanor:replace-base-image-2.11
Aug 7, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:release-ocm-2.11from
danmanor:replace-base-image-2.11

Conversation

@danmanor

@danmanor danmanor commented Aug 5, 2024 •

Copy link
Copy Markdown
Contributor

Backport of #6637

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Aug 5, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 5, 2024 •

Copy link
Copy Markdown

@danmanor: This pull request references MGMT-18313 which is a valid jira issue.

Details

In response to this:

Backport of #6637

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Aug 5, 2024
@openshift-ci
openshift-ci Bot requested review from oourfali and rccrdpccl August 5, 2024 14:12
@openshift-ci openshift-ci Bot added the kind/dependency-change Categorizes issue or PR as related to changing dependencies label Aug 5, 2024
@openshift-ci

openshift-ci Bot commented Aug 5, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: danmanor

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 5, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 5, 2024 •

Copy link
Copy Markdown

@danmanor: This pull request references MGMT-18313 which is a valid jira issue.

Details

In response to this:

Backport of #6637

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@codecov

codecov Bot commented Aug 5, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 68.50%. Comparing base (9509d78) to head (2a75a13).
Report is 1 commits behind head on release-ocm-2.11.

Additional details and impacted files

Impacted file tree graph

@@                Coverage Diff                @@
##           release-ocm-2.11    #6647   +/-   ##
=================================================
  Coverage             68.50%   68.50%           
=================================================
  Files                   247      247           
  Lines                 36334    36334           
=================================================
  Hits                  24889    24889           
  Misses                 9242     9242           
  Partials               2203     2203           

see 1 file with indirect coverage changes

@danmanor

danmanor commented Aug 5, 2024

Copy link
Copy Markdown
Contributor Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 5, 2024
@danmanor
danmanor force-pushed the replace-base-image-2.11 branch from c23329e to 1ebd949 Compare August 5, 2024 16:01
@danmanor

danmanor commented Aug 5, 2024

Copy link
Copy Markdown
Contributor Author

/unhold

@danmanor

danmanor commented Aug 5, 2024

Copy link
Copy Markdown
Contributor Author

/cc @adriengentil @eifrach

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 5, 2024
@openshift-ci
openshift-ci Bot requested review from adriengentil and eifrach August 5, 2024 16:07
@danmanor

danmanor commented Aug 6, 2024

Copy link
Copy Markdown
Contributor Author

/retest

@eifrach

eifrach commented Aug 6, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 6, 2024
@danmanor
danmanor force-pushed the replace-base-image-2.11 branch from 1ebd949 to 3e028ff Compare August 6, 2024 12:50
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Aug 6, 2024
@danmanor

danmanor commented Aug 6, 2024

Copy link
Copy Markdown
Contributor Author

/retest

@danmanor

danmanor commented Aug 6, 2024

Copy link
Copy Markdown
Contributor Author

/test images

@danmanor

danmanor commented Aug 6, 2024

Copy link
Copy Markdown
Contributor Author

/retest

@danmanor
danmanor force-pushed the replace-base-image-2.11 branch from 3e028ff to 2a75a13 Compare August 7, 2024 07:13
@eifrach

eifrach commented Aug 7, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 7, 2024
@openshift-ci

openshift-ci Bot commented Aug 7, 2024

Copy link
Copy Markdown
Contributor

@danmanor: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 957b236 into openshift:release-ocm-2.11 Aug 7, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. kind/dependency-change Categorizes issue or PR as related to changing dependencies lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants