Skip to content

[release-ocm-2.11] MGMT-18508: Allow converged from 4.17 hub - #6643

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:release-ocm-2.11from
carbonin:allow-converged-from-4.17-hub
Aug 2, 2024
Merged

openshift-merge-bot[bot] merged 2 commits into
openshift:release-ocm-2.11from
carbonin:allow-converged-from-4.17-hub

Conversation

@carbonin

@carbonin carbonin commented Aug 2, 2024

Copy link
Copy Markdown
Member

This PR backports #6533 and #6639

Together these PRs allow successful discovery of hosts using a 4.17 hub cluster and converged flow which will be required in 2.11 when 4.17 is released.

List all the issues related to this PR

Resolves https://issues.redhat.com/browse/MGMT-18508

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see [CONTRIBUTING] guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

adriengentil and others added 2 commits August 2, 2024 08:47
…ble (openshift#6533)

* MGMT-17478: Use ICC config secret for converged flow info when available

When available, read the ICC secret set by BMO in order to retrieve the
agent image, the agent URL(s), and the inspector URL(s).

This this change, we wil try to:
1/ Get the agent image from the user override
2/ if 1/ is not found, get the image from the ICC configuration
3/ if 2/ is not found, get the image from the HUB cluster
4/ if 3/ is not found, return the default image

* Add context to GetICCConfig method

* remove space

* refactor tiered logic

* Create GetImageArchitecture mehod

* fix lint

* refactor logs and getIronicAgentImageByRelease

* add infraenv in logger earlier

* make getICCConfig private
…nshift#6639)

* This allows ironic inspector URL to be missing in ICC config secret.

In 4.17 this is expected to be missing as the inspector service as been
removed.

If this URL is provided the agent will attempt to contact the inspector
service when it shouldn't causing the install to fail.

In earlier versions the secret will not be present so the controller
will continue to provide the inspector service URL as before.

Resolves https://issues.redhat.com/browse/OCPBUGS-37472

* Update image-customization-controller to release-4.16 branch

This includes a patch which removes the default for the inspector URL.
This is required because when deploying from a 4.17 hub the inspector
URL will not be present in the information on the cluster and we don't
want that URL to be set in the ignition.
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Aug 2, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 2, 2024 •

Copy link
Copy Markdown

@carbonin: This pull request references MGMT-18508 which is a valid jira issue.

Details

In response to this:

This PR backports #6533 and #6639

Together these PRs allow successful discovery of hosts using a 4.17 hub cluster and converged flow which will be required in 2.11 when 4.17 is released.

List all the issues related to this PR

Resolves https://issues.redhat.com/browse/MGMT-18508

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see [CONTRIBUTING] guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Aug 2, 2024
@openshift-ci
openshift-ci Bot requested a review from romfreiman August 2, 2024 12:52
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 2, 2024
@openshift-ci

openshift-ci Bot commented Aug 2, 2024

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: adriengentil, carbonin

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 2, 2024
@codecov

codecov Bot commented Aug 2, 2024 •

Copy link
Copy Markdown

Codecov Report

Attention: Patch coverage is 88.99083% with 12 lines in your changes missing coverage. Please review.

Project coverage is 68.51%. Comparing base (63039ec) to head (33de24e).
Report is 1 commits behind head on release-ocm-2.11.

Files Patch % Lines
...ler/controllers/preprovisioningimage_controller.go 86.66% 7 Missing and 5 partials ⚠️
Additional details and impacted files

Impacted file tree graph

@@                 Coverage Diff                  @@
##           release-ocm-2.11    #6643      +/-   ##
====================================================
+ Coverage             68.45%   68.51%   +0.05%     
====================================================
  Files                   247      247              
  Lines                 36267    36334      +67     
====================================================
+ Hits                  24827    24893      +66     
- Misses                 9238     9240       +2     
+ Partials               2202     2201       -1     
Files Coverage Δ
internal/controller/controllers/bmo_utils.go 70.83% <100.00%> (+9.01%) ⬆️
internal/oc/release.go 71.48% <100.00%> (+0.21%) ⬆️
...ler/controllers/preprovisioningimage_controller.go 82.38% <86.66%> (+1.38%) ⬆️

... and 3 files with indirect coverage changes

@openshift-ci

openshift-ci Bot commented Aug 2, 2024

Copy link
Copy Markdown
Contributor

@carbonin: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-merge-bot
openshift-merge-bot Bot merged commit 9509d78 into openshift:release-ocm-2.11 Aug 2, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants