Skip to content

MGMT-14975: Update MCE operator installation to use stable-2.3 channel - #5297

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
eliorerz:MGMT-14975_Update_MCE_operator_installation_to_use_stable-2.3_channel
Jun 15, 2023
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
eliorerz:MGMT-14975_Update_MCE_operator_installation_to_use_stable-2.3_channel

Conversation

@eliorerz

Copy link
Copy Markdown
Contributor

MCE 2.2 will not be available in OCP 4.14. Only MCE 2.3+, for that MCE operator subscription manifest channel need to be updated to stable-2.3.

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

/cc @gamli75

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jun 15, 2023
@openshift-ci-robot

openshift-ci-robot commented Jun 15, 2023 •

Copy link
Copy Markdown

@eliorerz: This pull request references MGMT-14975 which is a valid jira issue.

Details

In response to this:

MCE 2.2 will not be available in OCP 4.14. Only MCE 2.3+, for that MCE operator subscription manifest channel need to be updated to stable-2.3.

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

/cc @gamli75

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci
openshift-ci Bot requested a review from gamli75 June 15, 2023 16:36
@eliorerz

Copy link
Copy Markdown
Contributor Author

/test ?

@openshift-ci openshift-ci Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jun 15, 2023
@openshift-ci

openshift-ci Bot commented Jun 15, 2023

Copy link
Copy Markdown
Contributor

@eliorerz: The following commands are available to trigger required jobs:

  • /test e2e-agent-compact-ipv4
  • /test edge-assisted-operator-catalog-publish-verify
  • /test edge-ci-index
  • /test edge-e2e-ai-operator-ztp
  • /test edge-e2e-ai-operator-ztp-sno-day2-workers
  • /test edge-e2e-ai-operator-ztp-sno-day2-workers-late-binding
  • /test edge-e2e-metal-assisted
  • /test edge-e2e-metal-assisted-4-10
  • /test edge-e2e-metal-assisted-4-11
  • /test edge-e2e-metal-assisted-4-12
  • /test edge-e2e-metal-assisted-4-9
  • /test edge-e2e-metal-assisted-cnv
  • /test edge-e2e-metal-assisted-lvm
  • /test edge-e2e-metal-assisted-odf
  • /test edge-images
  • /test edge-lint
  • /test edge-subsystem-aws
  • /test edge-subsystem-kubeapi-aws
  • /test edge-unit-test
  • /test edge-verify-generated-code
  • /test images
  • /test mce-images

The following commands are available to trigger optional jobs:

  • /test e2e-agent-ha-dualstack
  • /test e2e-agent-sno-ipv6
  • /test edge-e2e-ai-operator-ztp-3masters
  • /test edge-e2e-ai-operator-ztp-capi
  • /test edge-e2e-ai-operator-ztp-compact-day2-masters
  • /test edge-e2e-ai-operator-ztp-compact-day2-workers
  • /test edge-e2e-ai-operator-ztp-disconnected
  • /test edge-e2e-ai-operator-ztp-hypershift-zero-nodes
  • /test edge-e2e-ai-operator-ztp-ipv4v6-3masters
  • /test edge-e2e-ai-operator-ztp-ipv4v6-3masters-ocp-411
  • /test edge-e2e-ai-operator-ztp-ipv4v6-sno
  • /test edge-e2e-ai-operator-ztp-ipv4v6-sno-ocp-411
  • /test edge-e2e-ai-operator-ztp-multiarch-3masters-ocp-411
  • /test edge-e2e-ai-operator-ztp-multiarch-sno-ocp-411
  • /test edge-e2e-ai-operator-ztp-node-labels
  • /test edge-e2e-ai-operator-ztp-sno-day2-masters
  • /test edge-e2e-ai-operator-ztp-sno-day2-workers-ignitionoverride
  • /test edge-e2e-metal-assisted-4-13
  • /test edge-e2e-metal-assisted-4-14
  • /test edge-e2e-metal-assisted-day2
  • /test edge-e2e-metal-assisted-day2-arm-workers
  • /test edge-e2e-metal-assisted-day2-single-node
  • /test edge-e2e-metal-assisted-ipv4v6
  • /test edge-e2e-metal-assisted-ipv6
  • /test edge-e2e-metal-assisted-kube-api-late-binding-single-node
  • /test edge-e2e-metal-assisted-kube-api-late-unbinding-ipv4-single-node
  • /test edge-e2e-metal-assisted-kube-api-net-suite
  • /test edge-e2e-metal-assisted-mce
  • /test edge-e2e-metal-assisted-mce-4-10
  • /test edge-e2e-metal-assisted-mce-sno
  • /test edge-e2e-metal-assisted-metallb
  • /test edge-e2e-metal-assisted-none
  • /test edge-e2e-metal-assisted-onprem
  • /test edge-e2e-metal-assisted-single-node
  • /test edge-e2e-metal-assisted-static-ip-suite
  • /test edge-e2e-metal-assisted-tang
  • /test edge-e2e-metal-assisted-tpmv2
  • /test edge-e2e-metal-assisted-upgrade-agent
  • /test edge-e2e-nutanix-assisted
  • /test edge-e2e-nutanix-assisted-2workers
  • /test edge-e2e-vsphere-assisted
  • /test edge-e2e-vsphere-assisted-4-12
  • /test edge-e2e-vsphere-assisted-4-13
  • /test edge-e2e-vsphere-assisted-umn
  • /test edge-push-pr-image
  • /test push-pr-image

Use /test all to run the following jobs that were automatically triggered:

  • pull-ci-openshift-assisted-service-master-e2e-agent-compact-ipv4
  • pull-ci-openshift-assisted-service-master-edge-ci-index
  • pull-ci-openshift-assisted-service-master-edge-e2e-ai-operator-ztp
  • pull-ci-openshift-assisted-service-master-edge-e2e-metal-assisted
  • pull-ci-openshift-assisted-service-master-edge-e2e-metal-assisted-cnv
  • pull-ci-openshift-assisted-service-master-edge-e2e-metal-assisted-lvm
  • pull-ci-openshift-assisted-service-master-edge-e2e-metal-assisted-mce
  • pull-ci-openshift-assisted-service-master-edge-e2e-metal-assisted-odf
  • pull-ci-openshift-assisted-service-master-edge-images
  • pull-ci-openshift-assisted-service-master-edge-lint
  • pull-ci-openshift-assisted-service-master-edge-subsystem-aws
  • pull-ci-openshift-assisted-service-master-edge-subsystem-kubeapi-aws
  • pull-ci-openshift-assisted-service-master-edge-unit-test
  • pull-ci-openshift-assisted-service-master-edge-verify-generated-code
  • pull-ci-openshift-assisted-service-master-images
  • pull-ci-openshift-assisted-service-master-mce-images
Details

In response to this:

/test ?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 15, 2023
@eliorerz

Copy link
Copy Markdown
Contributor Author

/test edge-e2e-metal-assisted-mce edge-e2e-metal-assisted-mce-4-10 edge-e2e-metal-assisted-mce-sno

@codecov

codecov Bot commented Jun 15, 2023 •

Copy link
Copy Markdown

Codecov Report

Merging #5297 (4d49d8b) into master (682809b) will increase coverage by 0.03%.
The diff coverage is 89.33%.

Additional details and impacted files

Impacted file tree graph

@@            Coverage Diff             @@
##           master    #5297      +/-   ##
==========================================
+ Coverage   67.50%   67.54%   +0.03%     
==========================================
  Files         221      221              
  Lines       33044    33108      +64     
==========================================
+ Hits        22307    22363      +56     
- Misses       8726     8733       +7     
- Partials     2011     2012       +1     
Impacted Files Coverage Δ
internal/common/conversions.go 6.06% <0.00%> (-1.09%) ⬇️
...nsion/v1beta1/agentclusterinstall_mutating_hook.go 58.25% <95.00%> (+2.93%) ⬆️
...ion/v1beta1/agentclusterinstall_validation_hook.go 87.40% <96.00%> (+2.10%) ⬆️

... and 2 files with indirect coverage changes

@gamli75

gamli75 commented Jun 15, 2023

Copy link
Copy Markdown
Contributor

/test e2e-agent-compact-ipv4

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jun 15, 2023
@openshift-ci

openshift-ci Bot commented Jun 15, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: eliorerz, gamli75

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jun 15, 2023

Copy link
Copy Markdown
Contributor

@eliorerz: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-robot
openshift-merge-robot merged commit db09755 into openshift:master Jun 15, 2023
@eliorerz
eliorerz deleted the MGMT-14975_Update_MCE_operator_installation_to_use_stable-2.3_channel branch June 16, 2023 05:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants