Skip to content

MGMT-14096: Don't wait for console if disabled - ACM 2.6 - #5067

Merged
openshift-merge-robot merged 1 commit into
openshift:release-ocm-2.6from
jhernand:dont_wait_for_console_before_marking_cluster_deployment_as_complete_acm_2_6
Mar 27, 2023
Merged

MGMT-14096: Don't wait for console if disabled - ACM 2.6#5067
openshift-merge-robot merged 1 commit into
openshift:release-ocm-2.6from
jhernand:dont_wait_for_console_before_marking_cluster_deployment_as_complete_acm_2_6

Conversation

@jhernand

@jhernand jhernand commented Mar 24, 2023

Copy link
Copy Markdown
Contributor

Currently the cluster deployment controller waits for the console to be available before marking the cluster deployment as completed. This fails when the console capability is disabled. To avoid that this patch changes the controller so that it doesn't wait when the console is disabled.

Note that this is a clone of MGMT-14096 (which in turn is a clone of OCPBUGS-8335) for ACM 2.6.

Related: https://issues.redhat.com/browse/OCPBUGS-8335
Related: https://issues.redhat.com/browse/MGMT-14096
Related: #5022

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Currently the cluster deployment controller waits for the console to be
available before marking the cluster deployment as completed. This fails
when the console capability is disabled. To avoid that this patch
changes the controller so that it doesn't wait when the console is
disabled.

Note that this is a clone of MGMT-14096 (which in turn is a clone of
OCPBUGS-8335) for ACM 2.6.

Related: https://issues.redhat.com/browse/OCPBUGS-8335
Related: https://issues.redhat.com/browse/MGMT-14096
Related: openshift#5022
Signed-off-by: Juan Hernandez <juan.hernandez@redhat.com>
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Mar 24, 2023
@openshift-ci-robot

openshift-ci-robot commented Mar 24, 2023

Copy link
Copy Markdown

@jhernand: This pull request references MGMT-14096 which is a valid jira issue.

Details

In response to this:

Currently the cluster deployment controller waits for the console to be available before marking the cluster deployment as completed. This fails when the console capability is disabled. To avoid that this patch changes the controller so that it doesn't wait when the console is disabled.

Note that this is a clone of MGMT-14096 (which in turn is a clone of OCPBUGS-8335) for ACM 2.6.

Related: https://issues.redhat.com/browse/OCPBUGS-8335
Related: https://issues.redhat.com/browse/MGMT-14096
Related: #5022

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

1 similar comment
@openshift-ci-robot

openshift-ci-robot commented Mar 24, 2023

Copy link
Copy Markdown

@jhernand: This pull request references MGMT-14096 which is a valid jira issue.

Details

In response to this:

Currently the cluster deployment controller waits for the console to be available before marking the cluster deployment as completed. This fails when the console capability is disabled. To avoid that this patch changes the controller so that it doesn't wait when the console is disabled.

Note that this is a clone of MGMT-14096 (which in turn is a clone of OCPBUGS-8335) for ACM 2.6.

Related: https://issues.redhat.com/browse/OCPBUGS-8335
Related: https://issues.redhat.com/browse/MGMT-14096
Related: #5022

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Mar 24, 2023
@openshift-ci
openshift-ci Bot requested review from eliorerz and omertuc March 24, 2023 14:37
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 24, 2023
@codecov

codecov Bot commented Mar 24, 2023

Copy link
Copy Markdown

Codecov Report

Merging #5067 (8210f89) into release-ocm-2.6 (e6407bf) will decrease coverage by 0.02%.
The diff coverage is 61.53%.

Additional details and impacted files

Impacted file tree graph

@@                 Coverage Diff                 @@
##           release-ocm-2.6    #5067      +/-   ##
===================================================
- Coverage            65.81%   65.79%   -0.02%     
===================================================
  Files                  189      189              
  Lines                26906    26914       +8     
===================================================
+ Hits                 17707    17708       +1     
- Misses                7575     7581       +6     
- Partials              1624     1625       +1     
Impacted Files Coverage Δ
internal/cluster/cluster.go 69.40% <0.00%> (-0.43%) ⬇️
internal/bminventory/inventory.go 69.85% <100.00%> (+0.02%) ⬆️

... and 1 file with indirect coverage changes

@openshift-ci

openshift-ci Bot commented Mar 24, 2023

Copy link
Copy Markdown
Contributor

@jhernand: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Mar 27, 2023
@openshift-ci

openshift-ci Bot commented Mar 27, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: filanov, jhernand

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 33aca9d into openshift:release-ocm-2.6 Mar 27, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants