MGMT-13963: Don't wait for console if disabled - #5040
Conversation
Currently the cluster deployment controller waits for the console to be available before marking the cluster deployment as completed. This fails when the console capability is disabled. To avoid that this patch changes the controller so that it doesn't wait when the console is disabled. Note that this is a clone of MGMT-13941 (which in turn is a clone of OCPBUGS-8335) for ACM 2.7. Related: https://issues.redhat.com/browse/OCPBUGS-8335 Related: https://issues.redhat.com/browse/MGMT-13941 Related: openshift#5022 Signed-off-by: Juan Hernandez <juan.hernandez@redhat.com>
|
@jhernand: This pull request references MGMT-13963 which is a valid jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
1 similar comment
|
@jhernand: This pull request references MGMT-13963 which is a valid jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
Codecov Report
Additional details and impacted files@@ Coverage Diff @@
## release-ocm-2.7 #5040 +/- ##
===================================================
- Coverage 67.14% 67.13% -0.01%
===================================================
Files 200 200
Lines 29867 29875 +8
===================================================
+ Hits 20055 20058 +3
- Misses 8022 8027 +5
Partials 1790 1790
|
|
@jhernand: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: filanov, jhernand The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Currently the cluster deployment controller waits for the console to be available before marking the cluster deployment as completed. This fails when the console capability is disabled. To avoid that this patch changes the controller so that it doesn't wait when the console is disabled.
Note that this is a clone of MGMT-13941 (which in turn is a clone of OCPBUGS-8335) for ACM 2.7.
Related: https://issues.redhat.com/browse/OCPBUGS-8335
Related: https://issues.redhat.com/browse/MGMT-13941
Related: #5022
List all the issues related to this PR
What environments does this code impact?
How was this code tested?
Checklist
docs, README, etc)Reviewers Checklist