Skip to content

MGMT-13657: Don't wait for console if it is disabled - ACM 2.6 - #4980

Merged
openshift-merge-robot merged 1 commit into
openshift:release-ocm-2.6from
jhernand:dont_wait_for_console_if_capability_is_disabled_acm_2_6_backport
Feb 16, 2023
Merged

openshift-merge-robot merged 1 commit into
openshift:release-ocm-2.6from
jhernand:dont_wait_for_console_if_capability_is_disabled_acm_2_6_backport

Conversation

@jhernand

@jhernand jhernand commented Feb 14, 2023

Copy link
Copy Markdown
Contributor

This is a backport of MGMT-12471 for ACM 2.6. It includes the changes from pull request #4594.

Currently the service keeps for each cluster a list of the operators that it will wait for before considering the cluster installed. This list includes the console operator. But since version 4.11 that can be disabled via the capabilities section of the installer configuration file. For example:

capabilities:
  baselineCapabilitySet: None
  additionalEnabledCapabilities:
  - baremetal

That will configure the cluster so that only the baremetal optional capability will be installed, which effectively disables the Console capability.

This patch changes the service so that it removes the console operator from that list when this kind of update is made to the installer configuration.

Related: https://issues.redhat.com/browse/MGMT-13657
Related: https://issues.redhat.com/browse/MGMT-12471
Related: #4594

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Tested with the included unit tests.

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Feb 14, 2023
@openshift-ci-robot

openshift-ci-robot commented Feb 14, 2023

Copy link
Copy Markdown

@jhernand: This pull request references MGMT-13657 which is a valid jira issue.

Details

In response to this:

This is a backport of MGMT-12471 for ACM 2.6. It includes the changes from pull request #4594.

Currently the service keeps for each cluster a list of the operators that it will wait for before considering the cluster installed. This list includes the console operator. But since version 4.11 that can be disabled via the capabilities section of the installer configuration file. For example:

capabilities:
 baselineCapabilitySet: None
 additionalEnabledCapabilities:
 - baremetal

That will configure the cluster so that only the baremetal optional capability will be installed, which effectively disables the Console capability.

This patch changes the service so that it removes the console operator from that list when this kind of update is made to the installer configuration.

Related: https://issues.redhat.com/browse/MGMT-13657
Related: https://issues.redhat.com/browse/MGMT-12471
Related: #4594

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Tested with the included unit tests.

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

1 similar comment
@openshift-ci-robot

openshift-ci-robot commented Feb 14, 2023

Copy link
Copy Markdown

@jhernand: This pull request references MGMT-13657 which is a valid jira issue.

Details

In response to this:

This is a backport of MGMT-12471 for ACM 2.6. It includes the changes from pull request #4594.

Currently the service keeps for each cluster a list of the operators that it will wait for before considering the cluster installed. This list includes the console operator. But since version 4.11 that can be disabled via the capabilities section of the installer configuration file. For example:

capabilities:
 baselineCapabilitySet: None
 additionalEnabledCapabilities:
 - baremetal

That will configure the cluster so that only the baremetal optional capability will be installed, which effectively disables the Console capability.

This patch changes the service so that it removes the console operator from that list when this kind of update is made to the installer configuration.

Related: https://issues.redhat.com/browse/MGMT-13657
Related: https://issues.redhat.com/browse/MGMT-12471
Related: #4594

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Tested with the included unit tests.

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci openshift-ci Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Feb 14, 2023
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 14, 2023
@codecov

codecov Bot commented Feb 14, 2023

Copy link
Copy Markdown

Codecov Report

Merging #4980 (66d9376) into release-ocm-2.6 (d4c02b2) will increase coverage by 0.05%.
The diff coverage is 86.36%.

Additional details and impacted files

Impacted file tree graph

@@                 Coverage Diff                 @@
##           release-ocm-2.6    #4980      +/-   ##
===================================================
+ Coverage            65.75%   65.80%   +0.05%     
===================================================
  Files                  189      189              
  Lines                26838    26904      +66     
===================================================
+ Hits                 17646    17705      +59     
- Misses                7571     7575       +4     
- Partials              1621     1624       +3     
Impacted Files Coverage Δ
internal/bminventory/inventory.go 69.82% <86.36%> (+0.37%) ⬆️

@jhernand jhernand changed the title MGMT-13657: Don't wait for console if it is disabled MGMT-13657: Don't wait for console if it is disabled - ACM 2.6 Feb 14, 2023
@jhernand
jhernand force-pushed the dont_wait_for_console_if_capability_is_disabled_acm_2_6_backport branch from bf66a0a to 748e936 Compare February 14, 2023 11:16
@jhernand

Copy link
Copy Markdown
Contributor Author

/retest

1 similar comment
@jhernand

Copy link
Copy Markdown
Contributor Author

/retest

This is a backport of MGMT-12471 for ACM 2.6. It includes the changes
from pull request openshift#4594.

Currently the service keeps for each cluster a list of the operators
that it will wait for before considering the cluster installed. This
list includes the console operator. But since version 4.11 that can be
disabled via the `capabilities` section of the installer configuration
file. For example:

```yaml
capabilities:
  baselineCapabilitySet: None
  additionalEnabledCapabilities:
  - baremetal
```

That will configure the cluster so that only the `baremetal` optional
capability will be installed, which effectively disables the `Console`
capability.

This patch changes the service so that it removes the console operator
from that list when this kind of update is made to the installer
configuration.

Related: https://issues.redhat.com/browse/MGMT-13657
Related: https://issues.redhat.com/browse/MGMT-12471
Related: openshift#4594
Signed-off-by: Juan Hernandez <juan.hernandez@redhat.com>
@jhernand
jhernand force-pushed the dont_wait_for_console_if_capability_is_disabled_acm_2_6_backport branch from 748e936 to 66d9376 Compare February 15, 2023 08:29
@openshift-ci

openshift-ci Bot commented Feb 15, 2023

Copy link
Copy Markdown
Contributor

@jhernand: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 16, 2023
@openshift-ci

openshift-ci Bot commented Feb 16, 2023

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: filanov, jhernand

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 8c094c7 into openshift:release-ocm-2.6 Feb 16, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants