Skip to content

TELCODOCS-831: Document capabilities exclusion - #4213

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
iranzo:capabilitiesexclude
Dec 12, 2022
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
iranzo:capabilitiesexclude

Conversation

@iranzo

@iranzo iranzo commented Aug 1, 2022

Copy link
Copy Markdown
Member

Document on OpenShift Capabilities customization to exclude components

List all the issues related to this PR

  • New Feature
  • Enhancement
  • Bug fix
  • Tests
  • Documentation
  • CI/CD

What environments does this code impact?

  • Automation (CI, tools, etc)
  • Cloud
  • Operator Managed Deployments
  • None

How was this code tested?

  • assisted-test-infra environment
  • dev-scripts environment
  • Reviewer's test appreciated
  • Waiting for CI to do a full test run
  • Manual (Elaborate on how it was tested)
  • No tests needed

Assignees

/cc @
/cc @

Checklist

  • Title and description added to both, commit and PR.
  • Relevant issues have been associated (see CONTRIBUTING guide)
  • Reviewers have been listed
  • This change does not require a documentation update (docstring, docs, README, etc)
  • Does this change include unit-tests (note that code changes require unit-tests)

Reviewers Checklist

  • Are the title and description (in both PR and commit) meaningful and clear?
  • Is there a bug required (and linked) for this change?
  • Should this PR be backported?

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Aug 1, 2022
@openshift-ci

openshift-ci Bot commented Aug 1, 2022

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci openshift-ci Bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Aug 1, 2022
@openshift-ci
openshift-ci Bot requested review from romfreiman and ybettan August 1, 2022 08:42
@codecov

codecov Bot commented Aug 26, 2022

Copy link
Copy Markdown

Codecov Report

Merging #4213 (5978982) into master (d7eb5d6) will decrease coverage by 1.03%.
The diff coverage is n/a.

❗ Current head 5978982 differs from pull request most recent head 13665ba. Consider uploading reports for the commit 13665ba to get more accurate results

Additional details and impacted files

Impacted file tree graph

@@            Coverage Diff             @@
##           master    #4213      +/-   ##
==========================================
- Coverage   67.60%   66.57%   -1.04%     
==========================================
  Files         202      195       -7     
  Lines       29913    27130    -2783     
==========================================
- Hits        20223    18061    -2162     
+ Misses       7890     7439     -451     
+ Partials     1800     1630     -170     
Impacted Files Coverage Δ
internal/common/common_unitest_db.go 0.00% <0.00%> (-18.15%) ⬇️
internal/controller/controllers/agent_reclaimer.go 85.93% <0.00%> (-10.99%) ⬇️
internal/network/utils.go 26.81% <0.00%> (-10.52%) ⬇️
pkg/staticnetworkconfig/generator.go 18.58% <0.00%> (-9.42%) ⬇️
internal/common/db.go 0.00% <0.00%> (-8.38%) ⬇️
internal/host/common.go 85.00% <0.00%> (-3.89%) ⬇️
pkg/s3wrapper/client.go 16.59% <0.00%> (-3.15%) ⬇️
internal/host/validator.go 78.50% <0.00%> (-2.94%) ⬇️
pkg/mirrorregistries/generator.go 62.22% <0.00%> (-2.37%) ⬇️
...nal/controller/controllers/bmh_agent_controller.go 67.73% <0.00%> (-1.95%) ⬇️
... and 66 more

Comment thread docs/user-guide/install-customization.md Outdated
Comment thread docs/user-guide/install-customization.md Outdated
@iranzo iranzo changed the title WIP: TELCODOCS-831: Document capabilities exclusion TELCODOCS-831: Document capabilities exclusion Sep 14, 2022
@iranzo
iranzo marked this pull request as ready for review September 14, 2022 07:33
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 14, 2022

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@iranzo where do you stand with this doc? Also, can you please try to remove console and see whether we still can manage to isntall?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Sure, I'll try to remove console and validate installation

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

any update?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I would say that @javipolo did test it as I was on leave and his feedback was added to the PR.

Can you confirm Javi?

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yep, I did try it and without the console component, AI got stuck on 'Finalizing' and ultimately failed because of timeout

@jhernand was working on fixing that issue

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

are we waiting for the fix in order to get this piece merged?
I believe we can merge it already, as it won't get published in the next few days?

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

That is addressed in #4594, reviews are welcome.

Comment thread docs/user-guide/install-customization.md Outdated
@openshift-ci

openshift-ci Bot commented Dec 6, 2022

Copy link
Copy Markdown
Contributor

@iranzo: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/edge-e2e-ai-operator-ztp ba652e806207604a808681a65d2c7582cbd5cb06 link true /test edge-e2e-ai-operator-ztp
ci/prow/edge-e2e-metal-assisted-ocs ba652e806207604a808681a65d2c7582cbd5cb06 link true /test edge-e2e-metal-assisted-ocs
ci/prow/edge-e2e-metal-assisted-odf ba652e806207604a808681a65d2c7582cbd5cb06 link true /test edge-e2e-metal-assisted-odf
ci/prow/edge-e2e-metal-assisted-cnv ba652e806207604a808681a65d2c7582cbd5cb06 link true /test edge-e2e-metal-assisted-cnv
ci/prow/edge-e2e-metal-assisted-lvm ba652e806207604a808681a65d2c7582cbd5cb06 link true /test edge-e2e-metal-assisted-lvm
ci/prow/e2e-agent-compact 40bf97db45f2d73a0b3272c91f7938e58d59d93a link true /test e2e-agent-compact

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-robot openshift-merge-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Dec 6, 2022
@openshift-merge-robot openshift-merge-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Dec 12, 2022
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Dec 12, 2022
@openshift-ci

openshift-ci Bot commented Dec 12, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: iranzo, javipolo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 12, 2022
@openshift-merge-robot
openshift-merge-robot merged commit c0b3526 into openshift:master Dec 12, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants