Skip to content

NO-ISSUE: Allow goup root to write into data dir - #258

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
adriengentil:fix-data-directory
Jul 8, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
adriengentil:fix-data-directory

Conversation

@adriengentil

Copy link
Copy Markdown
Contributor

When deployed on OCP, we need to allow the group "root" to write in
data dir because the UID will be randomly selected.

https://docs.openshift.com/container-platform/4.16/openshift_images/create-images.html#use-uid_create-images

When deployed on OCP, we need to allow the group "root" to write in
data dir because the UID will be randomly selected.

https://docs.openshift.com/container-platform/4.16/openshift_images/create-images.html#use-uid_create-images
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jul 8, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@adriengentil: This pull request explicitly references no jira issue.

Details

In response to this:

When deployed on OCP, we need to allow the group "root" to write in
data dir because the UID will be randomly selected.

https://docs.openshift.com/container-platform/4.16/openshift_images/create-images.html#use-uid_create-images

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jul 8, 2024
@codecov

codecov Bot commented Jul 8, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 65.51%. Comparing base (866a347) to head (cc24363).
Report is 2 commits behind head on main.

Additional details and impacted files

Impacted file tree graph

@@           Coverage Diff           @@
##             main     #258   +/-   ##
=======================================
  Coverage   65.51%   65.51%           
=======================================
  Files          24       24           
  Lines        1308     1308           
=======================================
  Hits          857      857           
  Misses        310      310           
  Partials      141      141           

@adriengentil

Copy link
Copy Markdown
Contributor Author

/retest

@adriengentil

Copy link
Copy Markdown
Contributor Author

/cc @pastequo @danmanor

@openshift-ci
openshift-ci Bot requested review from danmanor and pastequo July 8, 2024 14:21
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 8, 2024
@openshift-ci

openshift-ci Bot commented Jul 8, 2024

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: adriengentil, carbonin

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [adriengentil,carbonin]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 8, 2024
@openshift-ci

openshift-ci Bot commented Jul 8, 2024

Copy link
Copy Markdown

@adriengentil: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants