Skip to content

NO-ISSUE: Ensure data dir is writable by UID/GID - #243

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
adriengentil:fix-data-dir
Jul 3, 2024
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift:mainfrom
adriengentil:fix-data-dir

Conversation

@adriengentil

@adriengentil adriengentil commented Jul 2, 2024 •

Copy link
Copy Markdown
Contributor

Currently, the image service fails to start using podman with a
permission error:

{“file”:“/go/src/github.com/openshift/origin/main.go:132",“func”:“main.main.func1",“level”:“fatal”,“msg”:“Failed to populate image store: failed to download https://mirror.openshift.com/pub/openshift-v4/x86_64/dependencies/rhcos/4.14/4.14.15/rhcos-4.14.15-x86_64-live.x86_64.iso: unable to create a temp file for /data/rhcos-full-iso-4.14-414.92.202402130420-0-x86_64.iso: open /data/.rhcos-full-iso-4.14-414.92.202402130420-0-x86_64.iso271460763: permission denied\n”,“time”:“2024-07-02T15:46:45Z”}

This change ensures that /data directory is writable for the UID/GID
that will be used to execute the image service executable.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jul 2, 2024
@openshift-ci-robot

Copy link
Copy Markdown

@adriengentil: This pull request explicitly references no jira issue.

Details

In response to this:

Currently, the image service fails to start using podman with a
permission error:

{“file”:“/go/src/github.com/openshift/origin/main.go:132",“func”:“main.main.func1",“level”:“fatal”,“msg”:“Failed to populate image store: failed to download https://mirror.openshift.com/pub/openshift-v4/x86_64/dependencies/rhcos/4.14/4.14.15/rhcos-4.14.15-x86_64-live.x86_64.iso: unable to create a temp file for /data/rhcos-full-iso-4.14-414.92.202402130420-0-x86_64.iso: open /data/.rhcos-full-iso-4.14-414.92.202402130420-0-x86_64.iso271460763: permission denied\n”,“time”:“2024-07-02T15:46:45Z”}

This change ensures that /data directory is created with the UID/GID
that will be used to execute the image service executable.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 2, 2024
@openshift-ci
openshift-ci Bot requested review from pastequo and tsorya July 2, 2024 16:39
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 2, 2024
Currently, the image service fails to start using podman with a
permission error:

```
{“file”:“/go/src/github.com/openshift/origin/main.go:132",“func”:“main.main.func1",“level”:“fatal”,“msg”:“Failed to populate image store: failed to download https://mirror.openshift.com/pub/openshift-v4/x86_64/dependencies/rhcos/4.14/4.14.15/rhcos-4.14.15-x86_64-live.x86_64.iso: unable to create a temp file for /data/rhcos-full-iso-4.14-414.92.202402130420-0-x86_64.iso: open /data/.rhcos-full-iso-4.14-414.92.202402130420-0-x86_64.iso271460763: permission denied\n”,“time”:“2024-07-02T15:46:45Z”}
```

This change ensures that `/data` directory is writable for the UID/GID
that will be used to execute the image service executable.
@adriengentil

Copy link
Copy Markdown
Contributor Author

/retitle NO-ISSUE: Create data dir after changing UID/GID

@openshift-ci openshift-ci Bot changed the title WIP: NO-ISSUE: Create data dir after changing UID/GID NO-ISSUE: Create data dir after changing UID/GID Jul 3, 2024
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Jul 3, 2024
@adriengentil

Copy link
Copy Markdown
Contributor Author

/retitle NO-ISSUE: Ensure data dir is writable by UID/GID

@openshift-ci openshift-ci Bot changed the title NO-ISSUE: Create data dir after changing UID/GID NO-ISSUE: Ensure data dir is writable by UID/GID Jul 3, 2024
@pastequo

pastequo commented Jul 3, 2024

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 3, 2024
@openshift-ci

openshift-ci Bot commented Jul 3, 2024

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: adriengentil, pastequo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [adriengentil,pastequo]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD f8c1e51 and 2 for PR HEAD 1054fdc in total

@codecov

codecov Bot commented Jul 3, 2024 •

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 65.51%. Comparing base (9f5422b) to head (1054fdc).
Report is 3 commits behind head on main.

Additional details and impacted files

Impacted file tree graph

@@           Coverage Diff           @@
##             main     #243   +/-   ##
=======================================
  Coverage   65.51%   65.51%           
=======================================
  Files          24       24           
  Lines        1308     1308           
=======================================
  Hits          857      857           
  Misses        310      310           
  Partials      141      141           

@adriengentil

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci

openshift-ci Bot commented Jul 3, 2024

Copy link
Copy Markdown

@adriengentil: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants