Skip to content

CFE-1114: UPSTREAM: <carry>: Bump kubernetes.core to v2.4.2 - #6

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
chiragkyal:update-ansile-collection
Aug 26, 2024
Merged

CFE-1114: UPSTREAM: <carry>: Bump kubernetes.core to v2.4.2#6
openshift-merge-bot[bot] merged 2 commits into
openshift:mainfrom
chiragkyal:update-ansile-collection

Conversation

@chiragkyal

@chiragkyal chiragkyal commented Jul 31, 2024

Copy link
Copy Markdown
Member

Description of the change:

  • Update openshift/release/ansible/requirements.yml file to use kubernetes.core v2.4.2
  • Run openshift/release/ansible/update-collections.sh script to update the vendored collections
  • Update openshift/ci/testdata/ansible/memcached-operator/requirements.yml to use kubernetes.core v2.4.2 in order to match with the downstream image dependencies.

Motivation for the change:
In favour of

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Jul 31, 2024
@chiragkyal chiragkyal changed the title Bump kubernetes.core to v2.4.2 CFE-1114: Bump kubernetes.core to v2.4.2 Aug 14, 2024
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Aug 14, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 14, 2024

Copy link
Copy Markdown

@chiragkyal: This pull request references CFE-1114 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.18.0" version, but no target version was set.

Details

In response to this:

Description of the change:

  • Update openshift/release/ansible/requirements.yml file to use kubernetes.core v2.4.2
  • Run openshift/release/ansible/update-collections.sh script to update the vendored collections

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@chiragkyal
chiragkyal marked this pull request as ready for review August 14, 2024 08:28
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Aug 14, 2024
@chiragkyal

Copy link
Copy Markdown
Member Author

/hold
for #12

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 14, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 14, 2024

Copy link
Copy Markdown

@chiragkyal: This pull request references CFE-1114 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.18.0" version, but no target version was set.

Details

In response to this:

Description of the change:

  • Update openshift/release/ansible/requirements.yml file to use kubernetes.core v2.4.2
  • Run openshift/release/ansible/update-collections.sh script to update the vendored collections

Motivation for the change:
In favour of openshift/ocp-release-operator-sdk#374

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci-robot

openshift-ci-robot commented Aug 14, 2024

Copy link
Copy Markdown

@chiragkyal: This pull request references CFE-1114 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.18.0" version, but no target version was set.

Details

In response to this:

Description of the change:

  • Update openshift/release/ansible/requirements.yml file to use kubernetes.core v2.4.2
  • Run openshift/release/ansible/update-collections.sh script to update the vendored collections

Motivation for the change:
In favour of

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@arkadeepsen

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 14, 2024
@chiragkyal

Copy link
Copy Markdown
Member Author

We need to address #12 (comment) here

@arkadeepsen

Copy link
Copy Markdown
Member

/lgtm cancel
Until #6 (comment) is addressed.

@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Aug 14, 2024
@chiragkyal

Copy link
Copy Markdown
Member Author

/test all

1 similar comment
@chiragkyal

Copy link
Copy Markdown
Member Author

/test all

@chiragkyal chiragkyal changed the title CFE-1114: Bump kubernetes.core to v2.4.2 CFE-1114: UPSTREAM: <carry>: Bump kubernetes.core to v2.4.2 Aug 22, 2024
Signed-off-by: chiragkyal <ckyal@redhat.com>
update openshift/ci/testdata/ansible/memcached-operator/requirements.yml
to use kubernetes.core v2.4.2 in order to match with downstream image

Signed-off-by: chiragkyal <ckyal@redhat.com>
@chiragkyal
chiragkyal force-pushed the update-ansile-collection branch from 61fdb33 to d80064f Compare August 26, 2024 07:01
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 26, 2024
@openshift-ci

openshift-ci Bot commented Aug 26, 2024

Copy link
Copy Markdown

@chiragkyal: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-ci-robot

openshift-ci-robot commented Aug 26, 2024

Copy link
Copy Markdown

@chiragkyal: This pull request references CFE-1114 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.18.0" version, but no target version was set.

Details

In response to this:

Description of the change:

  • Update openshift/release/ansible/requirements.yml file to use kubernetes.core v2.4.2
  • Run openshift/release/ansible/update-collections.sh script to update the vendored collections
  • Update openshift/ci/testdata/ansible/memcached-operator/requirements.yml to use kubernetes.core v2.4.2 in order to match with the downstream image dependencies.

Motivation for the change:
In favour of

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@TrilokGeer

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 26, 2024
@openshift-ci

openshift-ci Bot commented Aug 26, 2024

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: chiragkyal, TrilokGeer

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [TrilokGeer,chiragkyal]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@KeenonLee KeenonLee left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/label qe-approved

@openshift-ci openshift-ci Bot added the qe-approved Signifies that QE has signed off on this PR label Aug 26, 2024
@openshift-ci-robot

openshift-ci-robot commented Aug 26, 2024

Copy link
Copy Markdown

@chiragkyal: This pull request references CFE-1114 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "4.18.0" version, but no target version was set.

Details

In response to this:

Description of the change:

  • Update openshift/release/ansible/requirements.yml file to use kubernetes.core v2.4.2
  • Run openshift/release/ansible/update-collections.sh script to update the vendored collections
  • Update openshift/ci/testdata/ansible/memcached-operator/requirements.yml to use kubernetes.core v2.4.2 in order to match with the downstream image dependencies.

Motivation for the change:
In favour of

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@chiragkyal

Copy link
Copy Markdown
Member Author

/unhold
#6 (comment) is addressed

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 26, 2024
@openshift-merge-bot
openshift-merge-bot Bot merged commit c07954c into openshift:main Aug 26, 2024
@arkadeepsen

Copy link
Copy Markdown
Member

/cherrypick release-4.17

@openshift-cherrypick-robot

Copy link
Copy Markdown
Contributor

@arkadeepsen: new pull request created: #18

Details

In response to this:

/cherrypick release-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@arkadeepsen

Copy link
Copy Markdown
Member

/cherrypick release-4.17

@openshift-cherrypick-robot

Copy link
Copy Markdown
Contributor

@arkadeepsen: new pull request could not be created: failed to create pull request against openshift/ansible-operator-plugins#release-4.17 from head openshift-cherrypick-robot:cherry-pick-6-to-release-4.17: status code 422 not one of [201], body: {"message":"Validation Failed","errors":[{"resource":"PullRequest","code":"custom","message":"A pull request already exists for openshift-cherrypick-robot:cherry-pick-6-to-release-4.17."}],"documentation_url":"https://docs.github.com/rest/pulls/pulls#create-a-pull-request","status":"422"}

Details

In response to this:

/cherrypick release-4.17

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-bot

Copy link
Copy Markdown

[ART PR BUILD NOTIFIER]

Distgit: openshift-enterprise-ansible-operator
This PR has been included in build openshift-enterprise-ansible-operator-container-v4.18.0-202408261944.p0.gc07954c.assembly.stream.el8.
All builds following this will include this PR.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. qe-approved Signifies that QE has signed off on this PR

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants