Skip to content

OCPBUGS-85432: Increase resources for lvm in agent iso-no-registry - #1891

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
bfournie:lvm-iso-no-registry
May 12, 2026
Merged

OCPBUGS-85432: Increase resources for lvm in agent iso-no-registry#1891
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
bfournie:lvm-iso-no-registry

Conversation

@bfournie

Copy link
Copy Markdown
Contributor

By default the lvm-operator is now included as a dependency for CNV in the iso-no-registry test. The MASTER_VCPU must be increased to 10 in order for the master roles to be correctly assigned. In addition an additional disk must be created, as LVM requires at least one non-installation HDD/SSD disk on the host.

By default the lvm-operator is now included as a dependency for CNV
in the iso-no-registry test. The MASTER_VCPU must be increased to
10 in order for the master roles to be correctly assigned. In
addition an additional disk must be created, as LVM requires at
least one non-installation HDD/SSD disk on the host.
@openshift-ci

openshift-ci Bot commented May 12, 2026

Copy link
Copy Markdown

@bfournie: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-metal-ovn-arbiter da00d2e link false /test e2e-metal-ovn-arbiter

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@andfasano

Copy link
Copy Markdown
Member

/approve

@openshift-ci

openshift-ci Bot commented May 12, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: andfasano

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 12, 2026
@bfournie

Copy link
Copy Markdown
Contributor Author

Tested iso-no-registry with openshift/assisted-service#10285 and confirmed successful installation.

@pawanpinjarkar

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 12, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 574ecf5 into openshift-metal3:master May 12, 2026
17 of 18 checks passed
@bfournie
bfournie deleted the lvm-iso-no-registry branch May 12, 2026 19:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants