Skip to content

AGENT-1427: map NoRegistryClusterInstall tests to Installer / Agent based installation - #767

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:mainfrom
andfasano:add-abi-cr
Jul 2, 2026
Merged

AGENT-1427: map NoRegistryClusterInstall tests to Installer / Agent based installation#767
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:mainfrom
andfasano:add-abi-cr

Conversation

@andfasano

Copy link
Copy Markdown
Contributor

This patch is required to show the include the NoRegistryClusterInstall feature tests into their own row under Installer / Agent Based Installation in the Component Readiness page

…ation

add a matcher for [Feature:NoRegistryClusterInstall] tests and an OWNERS file for the Agent-based Installer team.
@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Jul 2, 2026
@openshift-ci-robot

openshift-ci-robot commented Jul 2, 2026

Copy link
Copy Markdown

@andfasano: This pull request references AGENT-1427 which is a valid jira issue.

Warning: The referenced jira issue has an invalid target version for the target branch this PR targets: expected the story to target the "5.0.0" version, but no target version was set.

Details

In response to this:

This patch is required to show the include the NoRegistryClusterInstall feature tests into their own row under Installer / Agent Based Installation in the Component Readiness page

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@andfasano

Copy link
Copy Markdown
Contributor Author

cc @pamoedom

@openshift-ci
openshift-ci Bot requested review from deads2k and smg247 July 2, 2026 13:03
@bfournie

bfournie commented Jul 2, 2026

Copy link
Copy Markdown

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 2, 2026
@stbenjam

stbenjam commented Jul 2, 2026

Copy link
Copy Markdown
Member

/approve

@openshift-ci

openshift-ci Bot commented Jul 2, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: andfasano, bfournie, stbenjam

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 2, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 08c938b into openshift-eng:main Jul 2, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants