Skip to content

add note informing users of env var updates - #10819

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
juanvallejo:jvallejo/add-clarification-flag-handling
Nov 5, 2018
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
juanvallejo:jvallejo/add-clarification-flag-handling

Conversation

@juanvallejo

Copy link
Copy Markdown
Contributor

Add note clarifying environment variables and flag handling by plugins under the new mechanism.

cc @soltysh

@k8s-ci-robot k8s-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Nov 2, 2018
@juanvallejo

Copy link
Copy Markdown
Contributor Author

cc @seans3

@k8sio-netlify-preview-bot

Copy link
Copy Markdown
Collaborator

Deploy preview for kubernetes-io-master-staging ready!

Built with commit 7c94bdf

https://deploy-preview-10819--kubernetes-io-master-staging.netlify.com

1 similar comment
@k8sio-netlify-preview-bot

Copy link
Copy Markdown
Collaborator

Deploy preview for kubernetes-io-master-staging ready!

Built with commit 7c94bdf

https://deploy-preview-10819--kubernetes-io-master-staging.netlify.com

@juanvallejo

Copy link
Copy Markdown
Contributor Author

@seans3

seans3 commented Nov 2, 2018

Copy link
Copy Markdown
Contributor

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Nov 2, 2018

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@cody-clark

Copy link
Copy Markdown
Contributor

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cody-clark

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 2, 2018
@xiangpengzhao

Copy link
Copy Markdown
Contributor

/check-cla

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Nov 5, 2018
@k8s-ci-robot
k8s-ci-robot merged commit 795de19 into kubernetes:master Nov 5, 2018
@juanvallejo
juanvallejo deleted the jvallejo/add-clarification-flag-handling branch November 5, 2018 14:37
chenopis added a commit that referenced this pull request Nov 5, 2018
k8s-ci-robot pushed a commit that referenced this pull request Nov 5, 2018
* Revert "Update tutorial to work with Katacoda, No Install (#10797)"

This reverts commit a0b97a3.

* Revert "clean-up minikube toc (#10873)"

This reverts commit 18343eb.

* Revert "Tutorial cleanup: Delete K8s 101 and 201 (#10786)"

This reverts commit e0168fd.

* Revert "fix word err in Disruption (#10884)"

This reverts commit d01b76d.

* Revert "Fix typo (#10877)"

This reverts commit 1551868.

* Revert "add note informing users of env var updates (#10819)"

This reverts commit 795de19.

* Revert "Fix typo in word repositories (#10883)"

This reverts commit 6d7f27e.

* Revert "Add i18n strings for Korean translation (#10611)"

This reverts commit a9d8c54.

* Revert "Modified into appropriate words (#10768)"

This reverts commit 7ebdd66.

* Revert "Update tools.md (#10753)"

This reverts commit 2d8bc15.

* Revert "Enable reviews/approvers for localized site strings (#10811)"

This reverts commit 81b2562.
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants