Skip to content

🌱 Remove deprecated ClusterResourceSet feature gate - #12950

Merged
k8s-ci-robot merged 1 commit into
kubernetes-sigs:mainfrom
aman4433:remove-clusterresourceset-featuregate
Nov 10, 2025
Merged

k8s-ci-robot merged 1 commit into
kubernetes-sigs:mainfrom
aman4433:remove-clusterresourceset-featuregate

Conversation

@aman4433

@aman4433 aman4433 commented Nov 4, 2025

Copy link
Copy Markdown
Contributor

What this PR does / why we need it:
This PR removes the deprecated ClusterResourceSet feature gate, which has been GA since v1.11.

Ref: Remove previously deprecated code

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):
Task of #12771
Fixes #11114

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. do-not-merge/needs-area PR is missing an area label labels Nov 4, 2025
@k8s-ci-robot k8s-ci-robot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Nov 4, 2025
@aman4433
aman4433 force-pushed the remove-clusterresourceset-featuregate branch from 6477e1e to 3e358dc Compare November 4, 2025 13:12
@k8s-ci-robot k8s-ci-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Nov 4, 2025
@aman4433
aman4433 force-pushed the remove-clusterresourceset-featuregate branch from 3e358dc to fa597bf Compare November 4, 2025 13:33
@aman4433

aman4433 commented Nov 4, 2025

Copy link
Copy Markdown
Contributor Author

/test pull-cluster-api-test-main

@aman4433

aman4433 commented Nov 4, 2025

Copy link
Copy Markdown
Contributor Author

/test pull-cluster-api-e2e-blocking-main

Comment thread main.go
@aman4433
aman4433 force-pushed the remove-clusterresourceset-featuregate branch from fa597bf to 7bc675b Compare November 6, 2025 13:07
Comment thread internal/webhooks/clusterresourcesetbinding.go Outdated
Comment thread internal/webhooks/clusterresourceset.go Outdated
Comment thread feature/feature.go
@aman4433
aman4433 force-pushed the remove-clusterresourceset-featuregate branch from 7bc675b to 46de674 Compare November 9, 2025 17:34
@k8s-ci-robot

k8s-ci-robot commented Nov 9, 2025

Copy link
Copy Markdown
Contributor

@aman4433: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-cluster-api-apidiff-main 46de674 link false /test pull-cluster-api-apidiff-main

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@sbueringer

Copy link
Copy Markdown
Member

/test pull-cluster-api-e2e-main-gke

@sbueringer sbueringer added the area/clusterresourceset Issues or PRs related to clusterresourcesets label Nov 10, 2025
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/needs-area PR is missing an area label label Nov 10, 2025
@sbueringer

Copy link
Copy Markdown
Member

Thank you very much!!

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Nov 10, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: 9988271cb6f12ac6bdc71227ba132fa07588c367

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: sbueringer

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 10, 2025
@k8s-ci-robot
k8s-ci-robot merged commit 1477456 into kubernetes-sigs:main Nov 10, 2025
20 of 21 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.12 milestone Nov 10, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/clusterresourceset Issues or PRs related to clusterresourcesets cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Feature: ClusterResourceSet

3 participants