Skip to content

✨ Promote feature CRS to GA - #11365

Merged
k8s-ci-robot merged 4 commits into
kubernetes-sigs:mainfrom
mcbenjemaa:crs-ga
Jan 20, 2025
Merged

k8s-ci-robot merged 4 commits into
kubernetes-sigs:mainfrom
mcbenjemaa:crs-ga

Conversation

@mcbenjemaa

@mcbenjemaa mcbenjemaa commented Nov 1, 2024

Copy link
Copy Markdown
Member

What this PR does / why we need it:
Promote ClusterResourceSet to GA,
Make the CRS feature GA and enabled by default, and make the CRS feature clear in the API and documentation.
the code organization will be held in following PR after this is released

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):

Ref #11114

/area api
/area clusterresourceset
/area documentation

@k8s-ci-robot k8s-ci-robot added area/api Issues or PRs related to the APIs area/clusterresourceset Issues or PRs related to clusterresourcesets area/documentation Issues or PRs related to documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Nov 1, 2024
@k8s-ci-robot k8s-ci-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Nov 1, 2024
@mcbenjemaa

Copy link
Copy Markdown
Member Author

/cc fabriziopandini

@mcbenjemaa mcbenjemaa changed the title ✨ promote CRS to GA ✨ Promote CRS to GA Nov 1, 2024
@mcbenjemaa mcbenjemaa changed the title ✨ Promote CRS to GA ✨ Promote feature CRS to GA Nov 1, 2024
@mcbenjemaa
mcbenjemaa force-pushed the crs-ga branch 3 times, most recently from b5a503b to 653f6f0 Compare November 4, 2024 15:57
@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Nov 7, 2024
@fabriziopandini

Copy link
Copy Markdown
Member

@mcbenjemaa It would be nice if we can rebase this PR
@chrischdi @vincepri @enxebre What is your opinion about graduating CRS to GA? (follow up of the work to assess all the existing feature flags).

I'm personally +1, considering we did a good work to define scope and limitations of this feature in the last iteration on https://github.com/kubernetes-sigs/cluster-api/blob/main/docs/proposals/20200220-cluster-resource-set.md + we have https://github.com/kubernetes-sigs/cluster-api/blob/main/docs/proposals/20220712-cluster-api-addon-orchestration.md in case someone something more sophisticated.
(I'm also ok consider merging this after code freeze, if you need some more time)

@chrischdi

Copy link
Copy Markdown
Member

@mcbenjemaa It would be nice if we can rebase this PR @chrischdi @vincepri @enxebre What is your opinion about graduating CRS to GA? (follow up of the work to assess all the existing feature flags).

I'm personally +1, considering we did a good work to define scope and limitations of this feature in the last iteration on main/docs/proposals/20200220-cluster-resource-set.md + we have main/docs/proposals/20220712-cluster-api-addon-orchestration.md in case someone something more sophisticated. (I'm also ok consider merging this after code freeze, if you need some more time)

I'm good with getting this to GA.

@k8s-ci-robot k8s-ci-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Nov 25, 2024

@chrischdi chrischdi left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Nov 25, 2024
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: f93d562e33f79fc162196a0d5d62e257a96a5ad4

@mcbenjemaa

Copy link
Copy Markdown
Member Author

So this should go to v1.9 or to the next one?

@fabriziopandini

Copy link
Copy Markdown
Member

We are not in rush, Let's move to 1.10
But let's bring this up in one of the next office hours, so we can hopefully get some traction + set a lazy consensus and try to get this out of the table soon

@k8s-ci-robot k8s-ci-robot removed the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Dec 5, 2024

@chrischdi chrischdi left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

lgtm from my side, just nits

Comment thread exp/addons/api/v1beta1/clusterresourceset_types.go Outdated
Comment thread feature/feature.go Outdated
@sbueringer

Copy link
Copy Markdown
Member

@mcbenjemaa Do you have time to address @chrischdi's findings above? I think once these are fixed we can merge the PR

@mcbenjemaa

Copy link
Copy Markdown
Member Author

I'm ready

Comment thread feature/feature.go
@sbueringer sbueringer added the tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. label Jan 17, 2025
@sbueringer

Copy link
Copy Markdown
Member

Thank you!

/approve

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 17, 2025
@elmiko

elmiko commented Jan 17, 2025

Copy link
Copy Markdown
Contributor

changes lgtm, i'll leave the tag to another though as i'm just reviewing this for the first time.

@sbueringer

Copy link
Copy Markdown
Member

/assign @chrischdi @fabriziopandini

@fabriziopandini

Copy link
Copy Markdown
Member

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 20, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: d09167c3b9fd8c8d51158487ef9cb026b111b7c0

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: fabriziopandini, sbueringer

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [fabriziopandini,sbueringer]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/api Issues or PRs related to the APIs area/clusterresourceset Issues or PRs related to clusterresourcesets area/documentation Issues or PRs related to documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants