Skip to content

Force e2e integration pass before merging PRs - #364

Merged
k8s-ci-robot merged 1 commit into
kubernetes-sigs:mainfrom
nunnatsa:fix-protection-branch
Feb 3, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes-sigs:mainfrom
nunnatsa:fix-protection-branch

Conversation

@nunnatsa

Copy link
Copy Markdown
Contributor

What this PR does / why we need it:

The branch protection is managed from the
https://github.com/kubernetes/test-infra repository. It currently requires that the "Integration" job will pass. But since we now the integration is a matrics that creates two test lanes, then the "integration" rule is never satisfied.

This PR renames the original "integration" job, to "integration-tests", and adds new rollout job with the original name of "integration", to pass only if all the jobs defined by the matrix passed.

Now the existing "Integration" branch protection rule should work.

Release notes:

Nonw

The branch protection is managed from the
https://github.com/kubernetes/test-infra repository. It currently
requires that the "Integration" job will pass. But since we now the
integration is a matrics that creates two test lanes, then the
"integration" rule is never satisfied.

This PR renames the original "integration" job, to "integration-tests",
and adds new rollout job with the original name of "integration", to
pass only if all the jobs defined by the matrix passed.

Now the existing "Integration" branch protection rule should work.

Signed-off-by: Nahshon Unna Tsameret <nahsh.ut@gmail.com>
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: nunnatsa

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot requested review from qinqon and rmohr January 11, 2026 13:23
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. approved Indicates a PR has been approved by an approver from all required OWNERS files. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Jan 11, 2026
@nunnatsa

Copy link
Copy Markdown
Contributor Author

/ok-to-test

@k8s-ci-robot k8s-ci-robot added the ok-to-test Indicates a non-member PR verified by an org member that is safe to test. label Jan 11, 2026
@coveralls

Copy link
Copy Markdown

Pull Request Test Coverage Report for Build 20895850610

Details

  • 0 of 0 changed or added relevant lines in 0 files are covered.
  • No unchanged relevant lines lost coverage.
  • Overall coverage remained the same at 37.559%

Totals Coverage Status
Change from base Build 20809211765: 0.0%
Covered Lines: 1028
Relevant Lines: 2737

💛 - Coveralls

@damdo

damdo commented Jan 24, 2026

Copy link
Copy Markdown
Member

/release-note-none

@agradouski

Copy link
Copy Markdown
Contributor

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 3, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 91f8412 into kubernetes-sigs:main Feb 3, 2026
21 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants