Skip to content

Revert moving to claster-api v1beta2 - #363

Merged
k8s-ci-robot merged 1 commit into
kubernetes-sigs:release-0.11from
nunnatsa:r0.11-revert-v1-capi-api-to-beta1
Jan 13, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes-sigs:release-0.11from
nunnatsa:r0.11-revert-v1-capi-api-to-beta1

Conversation

@nunnatsa

@nunnatsa nunnatsa commented Jan 8, 2026

Copy link
Copy Markdown
Contributor

Note: Cherry-pick of #361

Going back to v1beta1, because some user can't move to cluster-api v1.11, and can't adopt v1beta2, yet

Going back to v1beta1

Signed-off-by: Nahshon Unna Tsameret <nahsh.ut@gmail.com>
@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 8, 2026
@k8s-ci-robot k8s-ci-robot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jan 8, 2026
@coveralls

Copy link
Copy Markdown

Pull Request Test Coverage Report for Build 20811790319

Details

  • 24 of 87 (27.59%) changed or added relevant lines in 10 files are covered.
  • 10 unchanged lines in 5 files lost coverage.
  • Overall coverage decreased (-0.7%) to 37.559%

Changes Missing Coverage Covered Lines Changed/Added Lines %
api/v1alpha1/kubevirtcluster_types.go 0 2 0.0%
api/v1alpha1/kubevirtmachine_types.go 0 2 0.0%
pkg/context/machine_context.go 0 5 0.0%
pkg/context/cluster_context.go 0 7 0.0%
controllers/kubevirtcluster_controller.go 3 11 27.27%
controllers/kubevirtmachine_controller.go 20 29 68.97%
pkg/capiv1beta1/resources.go 0 9 0.0%
pkg/testing/common.go 0 10 0.0%
api/v1alpha1/zz_generated.deepcopy.go 0 11 0.0%
Files with Coverage Reduction New Missed Lines %
api/v1alpha1/zz_generated.deepcopy.go 1 0.0%
controllers/kubevirtcluster_controller.go 1 41.87%
pkg/capiv1beta1/resources.go 1 0.0%
pkg/context/machine_context.go 3 0.0%
pkg/context/cluster_context.go 4 0.0%
Totals Coverage Status
Change from base Build 20760752719: -0.7%
Covered Lines: 1028
Relevant Lines: 2737

💛 - Coveralls

@nunnatsa

nunnatsa commented Jan 8, 2026

Copy link
Copy Markdown
Contributor Author

/ok-to-test

@k8s-ci-robot k8s-ci-robot added the ok-to-test Indicates a non-member PR verified by an org member that is safe to test. label Jan 8, 2026
@qinqon

qinqon commented Jan 13, 2026

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 13, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: nunnatsa, qinqon

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit 51dae1b into kubernetes-sigs:release-0.11 Jan 13, 2026
20 of 21 checks passed
@nunnatsa
nunnatsa deleted the r0.11-revert-v1-capi-api-to-beta1 branch January 13, 2026 12:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants