Skip to content

Implement a cert watcher to use new certificates without restarting - #196

Merged
k8s-ci-robot merged 2 commits into
kubernetes-csi:mainfrom
black-dragon74:impl-certwatcher
Jan 19, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes-csi:mainfrom
black-dragon74:impl-certwatcher

Conversation

@black-dragon74

@black-dragon74 black-dragon74 commented Jan 8, 2026

Copy link
Copy Markdown
Contributor

What type of PR is this?
/kind feature

What this PR does / why we need it:
This PR implements a CertWatcher which is responsible for watching the TLS Cert and Key files.

It also sets a callback GetCertificate which if set in the gRPC TLS opts, will be called every time a certificate data is required, enabling hot reloads and removing the requirement to restart the sidecar.

Which issue(s) this PR fixes:

Fixes #152

Special notes for your reviewer:
NONE

Does this PR introduce a user-facing change?:

NONE

@k8s-ci-robot k8s-ci-robot added release-note-none Denotes a PR that doesn't merit a release note. kind/feature Categorizes issue or PR as related to a new feature. labels Jan 8, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Welcome @black-dragon74!

It looks like this is your first PR to kubernetes-csi/external-snapshot-metadata 🎉. Please refer to our pull request process documentation to help your PR have a smooth ride to approval.

You will be prompted by a bot to use commands during the review process. Do not be afraid to follow the prompts! It is okay to experiment. Here is the bot commands documentation.

You can also check if kubernetes-csi/external-snapshot-metadata has its own contribution guidelines.

You may want to refer to our testing guide if you run into trouble with your tests not passing.

If you are having difficulty getting your pull request seen, please follow the recommended escalation practices. Also, for tips and tricks in the contribution process you may want to read the Kubernetes contributor cheat sheet. We want to make sure your contribution gets all the attention it needs!

Thank you, and welcome to Kubernetes. 😃

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @black-dragon74. Thanks for your PR.

I'm waiting for a kubernetes-csi member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot k8s-ci-robot added needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jan 8, 2026

@Rakshith-R Rakshith-R left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/ok-to-test

@k8s-ci-robot k8s-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Jan 8, 2026
@xing-yang

Copy link
Copy Markdown
Contributor

/assign

@iPraveenParihar

Copy link
Copy Markdown
Contributor

Integration tests are failing because readinessProbe for csi-snapshot-metadata sidecar is failing.
Sent a fix kubernetes-csi/csi-driver-host-path#624

cc @carlbraganza @xing-yang

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Jan 14, 2026
@kaovilai

Copy link
Copy Markdown
Contributor

fyi kubernetes-csi/csi-driver-host-path#624 merged with v0.2.0 release which does not contain the http health probe from #190 .

@Rakshith-R

Rakshith-R commented Jan 14, 2026

Copy link
Copy Markdown
Contributor

fyi kubernetes-csi/csi-driver-host-path#624 merged with v0.2.0 release which does not contain the http health probe from #190 .

https://github.com/kubernetes-csi/external-snapshot-metadata/actions/runs/20814019470/job/60361307710?pr=196
The integration test overrides the sidecar image with the one built from this pr.
We'll get these changes in and do a v0.3.0 release and update in hostpath.

This patch adds a certificate watcher that hot reloads
any certificate updates.

This enhancement prevents the need to restart the sidecar to load
the new certificates.

Signed-off-by: Niraj Yadav <niryadav@redhat.com>
Signed-off-by: Niraj Yadav <niryadav@redhat.com>
@k8s-ci-robot k8s-ci-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Jan 19, 2026
@Rakshith-R

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 19, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: black-dragon74, Rakshith-R

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jan 19, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 0ecff54 into kubernetes-csi:main Jan 19, 2026
9 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/feature Categorizes issue or PR as related to a new feature. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesn't merit a release note. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Add cert Watcher to be able to use new certificates without a restart

6 participants