Skip to content

[release/11.0.1xx-preview6] Source code updates from microsoft/vstest - #7422

Merged
lewing merged 3 commits into
release/11.0.1xx-preview6from
darc-release/11.0.1xx-preview6-a98fcacd-0d3f-4981-a91e-ba7c0afcc0d9
Jun 26, 2026
Merged

[release/11.0.1xx-preview6] Source code updates from microsoft/vstest#7422
lewing merged 3 commits into
release/11.0.1xx-preview6from
darc-release/11.0.1xx-preview6-a98fcacd-0d3f-4981-a91e-ba7c0afcc0d9

Conversation

@dotnet-maestro

@dotnet-maestro dotnet-maestro Bot commented Jun 26, 2026

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/microsoft/vstest

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/microsoft/vstest:347cdb9131ce271827b95d83048159e6e6f11137..https://github.com/dotnet/dotnet:darc-release/11.0.1xx-preview6-a98fcacd-0d3f-4981-a91e-ba7c0afcc0d9

@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-policy-service
dotnet-policy-service Bot requested a review from a team June 26, 2026 13:59
@dotnet-maestro

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@lewing
lewing merged commit 1bff05b into release/11.0.1xx-preview6 Jun 26, 2026
16 checks passed
@lewing
lewing deleted the darc-release/11.0.1xx-preview6-a98fcacd-0d3f-4981-a91e-ba7c0afcc0d9 branch June 26, 2026 21:16
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant