Skip to content

[Starlight] Add scroll to top button - #25886

Merged
kodster28 merged 1 commit into
productionfrom
scroll-to-top
Oct 16, 2025
Merged

[Starlight] Add scroll to top button#25886
kodster28 merged 1 commit into
productionfrom
scroll-to-top

Conversation

@kodster28

Copy link
Copy Markdown
Collaborator

Summary

Adds scroll to top button from Starlight plugins.

image

@github-actions

Copy link
Copy Markdown
Contributor

This pull request requires reviews from CODEOWNERS as it changes files that match the following patterns:

Pattern Owners
*.ts @cloudflare/pcx-content-engineering, @kodster28
* @cloudflare/pcx-technical-writing

@github-actions

Copy link
Copy Markdown
Contributor

@kodster28
kodster28 merged commit 513a810 into production Oct 16, 2025
11 checks passed
@kodster28
kodster28 deleted the scroll-to-top branch October 16, 2025 19:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants