Skip to content

feat: Add kube-qps and kube-burst parameters. - #769

Merged
archlitchi merged 1 commit into
Project-HAMi:masterfrom
chaunceyjiang:qps
Mar 7, 2025
Merged

feat: Add kube-qps and kube-burst parameters.#769
archlitchi merged 1 commit into
Project-HAMi:masterfrom
chaunceyjiang:qps

Conversation

@chaunceyjiang

@chaunceyjiang chaunceyjiang commented Jan 2, 2025

Copy link
Copy Markdown
Collaborator

What type of PR is this?

/kind feature
What this PR does / why we need it:
dd kube-qps and kube-burst parameters
Which issue(s) this PR fixes:
Fixes #745

Special notes for your reviewer:

Depends on PR #681

Does this PR introduce a user-facing change?:

@codecov

codecov Bot commented Jan 2, 2025

Copy link
Copy Markdown

Codecov Report

Attention: Patch coverage is 41.37931% with 17 lines in your changes missing coverage. Please review.

Files with missing lines Patch % Lines
pkg/util/client/client.go 41.37% 14 Missing and 3 partials ⚠️
Flag Coverage Δ
unittests 64.33% <41.37%> (+17.35%) ⬆️

Flags with carried forward coverage won't be shown. Click here to find out more.

Files with missing lines Coverage Δ
pkg/device/nvidia/device.go 37.27% <ø> (-0.49%) ⬇️
pkg/util/client/client.go 44.68% <41.37%> (-9.49%) ⬇️

... and 10 files with indirect coverage changes

🚀 New features to boost your workflow:
  • Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@hami-robott

hami-robott Bot commented Mar 5, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: chaunceyjiang
Once this PR has been reviewed and has the lgtm label, please assign archlitchi for approval. For more information see the Kubernetes Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@chaunceyjiang

Copy link
Copy Markdown
Collaborator Author

/kind feature

Signed-off-by: chaunceyjiang <chaunceyjiang@gmail.com>
@archlitchi

Copy link
Copy Markdown
Member

/lgtm

@archlitchi
archlitchi merged commit 43282bf into Project-HAMi:master Mar 7, 2025
@ouyangluwei163 ouyangluwei163 mentioned this pull request Apr 8, 2025
45 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

scheduler: client-side throttling, not priority and fairness

2 participants