Skip to content

feat: move nodepool replace and patch from frontend to backend - #5141

Closed
Miguel Soriano (miguelsorianod) wants to merge 3 commits into
mainfrom
msoriano-move-nodepool-repl-and-patch-to-backend
Closed

Miguel Soriano (miguelsorianod) wants to merge 3 commits into
mainfrom
msoriano-move-nodepool-repl-and-patch-to-backend

Conversation

@miguelsorianod

@miguelsorianod Miguel Soriano (miguelsorianod) commented May 6, 2026

Copy link
Copy Markdown
Collaborator

This moves nodepool replace and patch from frontend to backend. Frontend now just creates the corresponding ARM Operation, a nodepool update operation, and it delegates all the actual nodepool replace and patch work to backend.

On backend now we check whether internalID is empty or not to assess
whether the nodepool patch request to Clusters Service has been
submitted. Once that occurs, we check the state.

We introduce a controller DispatchNodePoolUpdate that takes care of
making sure that a PATCH call is sent to CS when a nodepool replace
or patch has been issued through the Frontend resulting in a nodepool
update active operation.

@openshift-ci

openshift-ci Bot commented May 6, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

// that indicates that the nodepool is in "updating" state on CS side but that technically doesn't cover case 1 if somehow CS can end up in a
// 'updating' state not triggered by this controller.
logger.Info("dispatching PATCH node pool to Cluster Service", "cs_node_pool_href", csIDFromNodePool.String(), "node_pool_resource_id", nodePool.ID.String())
_, err = c.clustersServiceClient.UpdateNodePool(ctx, csIDFromNodePool, csNodePoolBuilder)

Copy link
Copy Markdown
Collaborator Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is potentially the biggest contention point in the PR.

@mbarnes Matthew Barnes (mbarnes) May 8, 2026

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yeah, you're gonna have to check CS error messages. See this for an example.

@mbarnes

Copy link
Copy Markdown
Collaborator

Maybe let's focus on #5142 first and get that right. I made some comments there that are also applicable here.

@miguelsorianod

Copy link
Copy Markdown
Collaborator Author

As an important insight, we should move delete before and/or at the same time than create because of #5142 (comment)

On backend now we check whether internalID is empty or not to assess
whether the nodepool create request to Clusters Service has been
submitted. Once that occurs, we check the state.

We introduce a controller DispatchNodePoolUpdate that takes care of
making sure that a PATCH call is sent to CS when a nodepool replace
or patch has been issued through the Frontend resulting in a nodepool
update active operation.

Although the controller has been introduced, for now it is not active
because on the frontend side we still send the InternalID in the active
operation and the controller does not do any action when the InternalID
has been set.

The controller attempts to be resilient to failures.
This commit moves nodepool replace and patch from frontend to
backend. Frontend now just creates the corresponding ARM Operation,
a nodepool update operation, and it delegates all the actual nodepool
replace and patch work to backend.
@miguelsorianod
Miguel Soriano (miguelsorianod) force-pushed the msoriano-move-nodepool-repl-and-patch-to-backend branch from eb7769e to ac341e2 Compare June 2, 2026 12:56
@openshift-ci

openshift-ci Bot commented Jun 2, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: miguelsorianod
Once this PR has been reviewed and has the lgtm label, please assign geoberle for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jun 15, 2026

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@miguelsorianod

Copy link
Copy Markdown
Collaborator Author

Superseded by #6067

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants