Skip to content

Update cincinatti interaction to use maestro for clusterID - #5045

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
cs-159-refine-cluster-uid
Apr 28, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
mainfrom
cs-159-refine-cluster-uid

Conversation

@deads2k

Copy link
Copy Markdown
Collaborator

This disconnects cluster-service from another piece of our controllers.

We should consider a backup flow that uses an value when missing and see if we can supply the value to cluster-service and generate it ourselves.

Addresses the first cause of the nodepoolversion controller retry in https://prow.ci.openshift.org/view/gs/test-platform-results/pr-logs/pull/Azure_ARO-HCP/5020/pull-ci-Azure-ARO-HCP-main-e2e-parallel/2048471908730540032

This disconnects cluster-service from another piece of our controllers.

We should consider a backup flow that uses an value when missing and see
if we can supply the value to cluster-service and generate it ourselves.
@deads2k
David Eads (deads2k) force-pushed the cs-159-refine-cluster-uid branch from 4c2a7c4 to 3d9d7f8 Compare April 28, 2026 13:59

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

I left a comment in the cluster desired version controller about using a default UUID one for the case of install when we move the cluster creation call to the backend, that will be handled in the dedicated PR by Jakob Gray (@JakobGray)

@openshift-ci

openshift-ci Bot commented Apr 28, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: deads2k, machi1990

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit 9365922 into main Apr 28, 2026
15 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the cs-159-refine-cluster-uid branch April 28, 2026 20:00
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants