Skip to content

Automated - Update component image digests - #4592

Merged
openshift-merge-bot[bot] merged 3 commits into
mainfrom
automated---update-component-image-digests
Mar 24, 2026
Merged

Automated - Update component image digests#4592
openshift-merge-bot[bot] merged 3 commits into
mainfrom
automated---update-component-image-digests

Conversation

@aro-hcp-robot

@aro-hcp-robot aro-hcp-robot Bot commented Mar 24, 2026

Copy link
Copy Markdown
Contributor

This automated PR updates ARO-HCP container image digests to the latest versions from registries.

Name Old Digest New Digest Tag Date Status
aro-hcp-exporter 38294123323d… f51a27f9b992… 4b91a3a 2026-03-24 11:25 updated
backplaneAPI bfbd903e800a… 9595652e0704… 09ad80b 2026-03-24 00:51 updated
hypershift e7748a72fa8e… 5c4ce9ac3f41… latest 2026-03-24 11:10 updated
arohcpbackend 1f02a6963d6e… 85deff1008fd… 4b91a3a 2026-03-24 11:27 updated
admin-api 01223f7c60e3… aa49815f5f10… 4b91a3a 2026-03-24 11:28 updated
sessiongate ce93e49b2a1c… 776f190c68a6… 4b91a3a 2026-03-24 11:27 updated
arohcpfrontend 78749a8731e7… 1e183eef45f1… 4b91a3a 2026-03-24 11:26 updated
maestro 8afde7f96125… 195db1fb826c… 8e59adbc040351456481cb12a2959e82d49ae432 2026-03-24 09:15 updated
acm-operator 12594107ba68… 57efa3ac2136… v2.15.2-434 2026-03-24 05:31 updated
acm-mce 9ada289f06b9… ba98af6c6095… v2.10.2-457 2026-03-24 11:22 updated
imageSync b9a1de42a37c… de93a4b3df98… 4b91a3a 2026-03-24 11:26 updated

Schedule: Monday through Friday at 2 AM UTC
Generated by: periodic-ci-Azure-ARO-HCP-main-image-updater-tooling
Generated at: 2026-03-24T12:28:13+0000

@openshift-ci

openshift-ci Bot commented Mar 24, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: aro-hcp-image-bumper[bot]
Once this PR has been reviewed and has the lgtm label, please assign geoberle for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@raelga

Copy link
Copy Markdown
Collaborator

/test e2e-parallel

@aro-hcp-robot
aro-hcp-robot Bot force-pushed the automated---update-component-image-digests branch from 25d2db7 to 4bcc7ca Compare March 24, 2026 12:28
@openshift-merge-bot
openshift-merge-bot Bot merged commit df64b94 into main Mar 24, 2026
17 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the automated---update-component-image-digests branch March 24, 2026 20:33
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant